)]}'
{"id":"openstack%2Fironic~801705","triplet_id":"openstack%2Fironic~master~Ib1145f9d34ad53966c2f4b24981316d4a8763340","project":"openstack/ironic","branch":"master","topic":"follow-up","hashtags":[],"change_id":"Ib1145f9d34ad53966c2f4b24981316d4a8763340","subject":"Update vendor_passthru subscriptions releasenote","status":"MERGED","created":"2021-07-22 06:19:37.000000000","updated":"2021-07-22 17:12:56.000000000","submitted":"2021-07-22 08:14:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801705-follow-up","meta_rev_id":"7f4175cb0860b861ffa77eecd651ebac6266229f","_number":801705,"virtual_id_number":801705,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 08:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-07-22 07:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-07-22 07:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-22 06:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 07:51:32.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-07-22 17:12:56.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"aff0bf37b1cd7c5f52bc9aa5033bc1d8f4c2dc21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-07-22 06:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26534762f0fa1bcf29e9014580e84b1f9a42579b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 06:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6864a1362e2c4747a97867ec00a032a1 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17af08a7b45b48358ae43e6abfcd0cb0 : SUCCESS in 15m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a9b0614e1014e929615ec908d9ccb60 : SUCCESS in 7m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c48b226561124e9d9a75ab5ae5d2186f : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5afe12f69f925ac4ac46bae2bff04cd82639d0d7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-22 07:51:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ntrivial change, single core approval","accounts_in_message":[],"_revision_number":1},{"id":"162cfee5ae585ee816cc185d9fbc228067e199f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 07:51:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7492cbbd70d808843859564c95d35901f24542b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 08:14:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff83ba95ea24aa396c80ea1065c37f8 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfc9bc0468d4d3e9b301619f456a512 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb033f118814aa7a96d3e79bdafca6d : SUCCESS in 7m 28s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7fb2342d98e64ee98cfda86fb251d290 : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"75d96ef558ab3203834a3dd3b305ddf0356157c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 08:14:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ad6ae1d1b86b04ee7e8b7737083cf2910c180f67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 08:16:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d113ce1cc84a678a2e7647d325d62d : SUCCESS in 1m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c4ceb89b5c14e38a395cc191be5e5d7 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7f4175cb0860b861ffa77eecd651ebac6266229f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-07-22 17:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/912 : SUCCESS in 5h 7m 41s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/910 : SUCCESS in 4h 8m 4s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/905 : SUCCESS in 3h 57m 19s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/913 : SUCCESS in 2h 53m 23s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1160 : SUCCESS in 2h 8m 6s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/918 : SUCCESS in 1h 55m 17s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1165 : SUCCESS in 1h 46m 26s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1203 : SUCCESS in 1h 3m 49s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1198 : SUCCESS in 1h 2m 38s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1200 : SUCCESS in 1h 1m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2417571bef6849c634a1e79c3f061cfee39efc8","revisions":{"a2417571bef6849c634a1e79c3f061cfee39efc8":{"kind":"REWORK","_number":1,"created":"2021-07-22 06:19:37.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/05/801705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/05/801705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/05/801705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/05/801705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/05/801705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/05/801705/1"}}},"commit":{"parents":[{"commit":"4bc5142df26c8279e1ab15703cbfba6849654c4b","subject":"Add vendor_passthru method for subscriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4bc5142df26c8279e1ab15703cbfba6849654c4b"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-07-22 06:17:50.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-07-22 06:17:50.000000000","tz":120},"subject":"Update vendor_passthru subscriptions releasenote","message":"Update vendor_passthru subscriptions releasenote\n\nThis is a follow-up to I5b7fa99b0ee64ccdc0f62d9686df655082db3665\n\nChange-Id: Ib1145f9d34ad53966c2f4b24981316d4a8763340\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a2417571bef6849c634a1e79c3f061cfee39efc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a2417571bef6849c634a1e79c3f061cfee39efc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
