)]}'
{"id":"openstack%2Fironic~802462","triplet_id":"openstack%2Fironic~stable%2Fvictoria~I20836d586edccfb8cd8fed1f3a89f1497ff96943","project":"openstack/ironic","branch":"stable/victoria","topic":"openstack-tox-driver-libs-stable/victoria","hashtags":[],"change_id":"I20836d586edccfb8cd8fed1f3a89f1497ff96943","subject":"Use openstack-tox for ironic-tox-unit-with-driver-libs","status":"MERGED","created":"2021-07-27 19:01:43.000000000","updated":"2021-07-28 09:09:47.000000000","submitted":"2021-07-28 08:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802462-openstack-tox-driver-libs-stable/victoria","cherry_pick_of_change":764445,"cherry_pick_of_patch_set":1,"meta_rev_id":"6ccaf2524ce657747d3cf8c2879420942274bb9a","_number":802462,"virtual_id_number":802462,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 08:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-07-28 07:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-07-28 07:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:01:43.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-07-27 20:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 09:09:47.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"050abd1bb7452cd19604f4e01512b235c3dc09e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-07-27 19:01:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b5330ae360906d9ea4ebfc8e2022d83028c6892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:47:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0695903ee99246b883e0e50122e4d330 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e60a9def9bb41498a35df01439df394 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022d3eb096d0428293dafad4d0988f90 : SUCCESS in 10m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175653aa1417458b99503d94265dba20 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/788a0aaad8854fb794b8f234df5b9aec : SUCCESS in 15m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c74a4aa869054f20bab70d5479d07027 : SUCCESS in 11m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/46fe24a73f0d4234ac820327346dca5d : SUCCESS in 59m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/eb37b38d94b74575a7725517d46e8667 : SUCCESS in 1h 12m 53s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/8b8595a9509c43deb7b998a8c02ed1b6 : SUCCESS in 31m 21s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5b963090cae74c77b090452cd3b11d1e : SUCCESS in 1h 03m 28s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a0a8a75f54e04e32843404fa3598de0d : SUCCESS in 52m 50s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/52a99a73b8fb4a7ab204ce70834d86ff : SUCCESS in 48m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/551a317715484f1d87c3d2a169b464cb : SUCCESS in 54m 54s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b542345af4d0456ba3b269c44a686edb : SUCCESS in 1h 17m 04s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3f6e08be3f1f46588bb9933dcd1fa00b : SUCCESS in 54m 48s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/54a0c14cc71d42a4a448fd739936c468 : SUCCESS in 1h 42m 14s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e7036f54e7d246a5a689040c2cd3aa62 : SUCCESS in 1h 23m 58s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ed3ab9ae9efb40188ae38922695723d5 : SUCCESS in 36m 01s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/199b664bbde245a58389e9d50a4bdcb3 : SUCCESS in 1h 02m 03s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/c4dec688112d474197d74b3e6d41ee6d : SUCCESS in 1h 14m 30s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b1cd0f47e57e4f9aaf3c3312300ff0e4 : SUCCESS in 5m 21s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4acb3f48ae2246eeb51f94849b991359 : SUCCESS in 50m 13s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/91813d29fc564567a035e427c2b0b180 : SUCCESS in 58m 35s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3624f02b29e54620a9ddbbfb24234ea8 : SUCCESS in 54m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/30fefd0b302f4cc4aa0b4030a4ad7042 : SUCCESS in 54m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f61b967af2f4339a0b29cb3634c8886 : SUCCESS in 25m 15s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/9a9ce9b4483344039d090fe408f0aa06 : SUCCESS in 46m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a15df7ac416d6f283679e22423391ba72d3de810","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-07-28 07:08:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6dff3d08ba929431fd1687264d5405e43da3e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 07:09:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7638f751646ff6fa2b81b7d7ff1c6694fecaa5e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:41:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd3d705dddd49b4a403a866ffcad202 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/035537f49fa349f0819db60e3ec2c181 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f6a08af6c6641d297b8e1295f2644b1 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f638291b8c4f7a831b9631aa976137 : SUCCESS in 17m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/323a00cd457a4f3b878ab85f53cdeac8 : SUCCESS in 10m 47s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bdd37de8f10546508a2ce1a5c81f9e38 : SUCCESS in 59m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4db2f155ffb048e4ab913de21d3bd1fb : SUCCESS in 1h 16m 51s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1c9d0d62381b41e2b87041a13e0fc3d8 : SUCCESS in 27m 08s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/f09395fc692b482cb2409ab78f7d8aa5 : SUCCESS in 1h 14m 56s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3e61351d7aa747ce867c99d8e14b7bb4 : SUCCESS in 49m 26s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2462fe7c49bd486abe446a03e04b6402 : SUCCESS in 55m 55s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/0d6fe5506b7a4bf3a6a4a31c6baeba59 : SUCCESS in 47m 53s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/321aa92d7c0647b6a99f0b8c69f2aee9 : SUCCESS in 1h 18m 48s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/93affe29e17746979e8836acc794f079 : SUCCESS in 52m 25s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3db90163d40c4d6bbbd5ac20b583b303 : SUCCESS in 1h 26m 35s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4dea29b36efe4c52ba5bc5ac19283fd1 : SUCCESS in 1h 24m 54s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/9866c7f1bf2e4acb83c0669b64a93c12 : SUCCESS in 47m 54s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/796bb2fa2f694f9daca379fbce5ce49e : SUCCESS in 49m 58s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/51e0ee8d3ae04d14931dd40dd939f4cd : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"701e40bb809dff3f5026c21c1dabb8b2807d2641","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:41:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a6c0ac4c1feada80c027d2d970593ab5dec765c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:45:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d20e31e13cd425e8eaa0b1a8cfc6cc5 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6ccaf2524ce657747d3cf8c2879420942274bb9a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-07-28 09:09:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_idrac_victoria https://stash.dellemc-community.org/logs/R6515_bios_idrac_victoria/51 : SUCCESS in 3h 13m 55s\n- 14G_bios_idrac_victoria https://stash.dellemc-community.org/logs/14G_bios_idrac_victoria/52 : SUCCESS in 2h 42m 6s\n- 13G_bios_idrac_victoria https://stash.dellemc-community.org/logs/13G_bios_idrac_victoria/52 : SUCCESS in 2h 37m 32s\n- 14G_uefi_redfish_victoria https://stash.dellemc-community.org/logs/14G_uefi_redfish_victoria/64 : SUCCESS in 2h 23m 26s\n- 14G_bios_redfish_victoria https://stash.dellemc-community.org/logs/14G_bios_redfish_victoria/51 : SUCCESS in 2h 22m 21s\n- 13G_bios_ipmi_victoria https://stash.dellemc-community.org/logs/13G_bios_ipmi_victoria/51 : SUCCESS in 2h 4m 0s\n- R7515_uefi_redfish_victoria https://stash.dellemc-community.org/logs/R7515_uefi_redfish_victoria/60 : SUCCESS in 2h 2m 4s\n- 14G_uefi_idrac_victoria https://stash.dellemc-community.org/logs/14G_uefi_idrac_victoria/62 : SUCCESS in 1h 58m 43s\n- XE2420_uefi_redfish_victoria https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_victoria/60 : SUCCESS in 1h 20m 42s\n- 14G_uefi_ipmi_victoria https://stash.dellemc-community.org/logs/14G_uefi_ipmi_victoria/61 : SUCCESS in 1h 8m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2df5dc42a830448e850955e738a63f41dc3541c5","revisions":{"2df5dc42a830448e850955e738a63f41dc3541c5":{"kind":"REWORK","_number":1,"created":"2021-07-27 19:01:43.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/62/802462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/62/802462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/62/802462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/62/802462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/62/802462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/62/802462/1"}}},"commit":{"parents":[{"commit":"d09a158cce9f71e94369778dab0d7da157f2bafd","subject":"Fix iPXE docs: snponly is not always available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d09a158cce9f71e94369778dab0d7da157f2bafd"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-11-27 11:20:53.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-07-27 19:01:43.000000000","tz":0},"subject":"Use openstack-tox for ironic-tox-unit-with-driver-libs","message":"Use openstack-tox for ironic-tox-unit-with-driver-libs\n\nAll the tox jobs are based on openstack-tox, we should convert\nironic-tox-unit-with-driver-libs too.\n\nChange-Id: I20836d586edccfb8cd8fed1f3a89f1497ff96943\n(cherry picked from commit 475af371ddaa8a6d381c6b73193fedab0c058c6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2df5dc42a830448e850955e738a63f41dc3541c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2df5dc42a830448e850955e738a63f41dc3541c5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
