)]}'
{"id":"openstack%2Fironic~806751","triplet_id":"openstack%2Fironic~master~I1b334fd820527415db21e301ee131adb67710b58","project":"openstack/ironic","branch":"master","topic":"bp/keylime-attestation-interface","attention_set":{"32454":{"account":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"last_update":"2021-10-26 18:56:41.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1b334fd820527415db21e301ee131adb67710b58","subject":"Keylime attestation interface","status":"NEW","created":"2021-08-31 13:47:45.000000000","updated":"2021-10-26 19:21:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":29,"unresolved_comment_count":22,"has_review_started":true,"meta_rev_id":"041481ceb47988edded9a3921d5996187f4d4260","_number":806751,"virtual_id_number":806751,"owner":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-26 19:05:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 14:24:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 14:35:20.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-08-31 14:42:23.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"CC"},{"updated":"2021-08-31 14:46:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-08-31 16:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 10:53:10.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-09-20 18:44:24.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"39f3e3b293e432e784a789df23149fb44da8a6ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"date":"2021-08-31 13:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eee5979fb54e77d6abdb7a5bce8858fc02d59412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:24:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b1a94ffe74e4739867ad1ad944a6500 : FAILURE in 31m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1739d30e912d4dc4bf6daf0af78abc82 : FAILURE in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9464ef154534e4ce9824b267cc68d7dcdbb45597","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-08-31 14:35:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/67efc43/67efc43.html : FAILURE in 8m 00s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/5b5b90f/5b5b90f.html : FAILURE in 8m 09s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/49178d9/49178d9.html : FAILURE in 9m 19s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/bca7708/bca7708.html : FAILURE in 9m 51s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/6fd75ed/6fd75ed.html : FAILURE in 9m 29s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/1b14fc4/1b14fc4.html : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3256d5ba6d6118ca8f4594d20118226039fe7fb4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-08-31 14:42:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2d0ef10b592a95de7538046f959bb1e99118262","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-08-31 14:46:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(21 comments)\n\nGood start, a number of small items, but nothing insurmountable :)","accounts_in_message":[],"_revision_number":1},{"id":"0ff5dd7adb4da85d732f040285fc3c45c5f95bd6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-08-31 14:47:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80735d7590ca661699ab589c31a43cb0317e1b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:11:39.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e82295abab74045868578c7d1ff1bd8 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1483346484d41f8be0e163d5e29a00b : FAILURE in 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91efe9d8b1a349be84c021b3f2f50c38 : FAILURE in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd45105b13241da83c22f40a39d81c2 : FAILURE in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49b27e6fd3d645f7843e96cff08ea32e : FAILURE in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10f585a77dd549ecbb7750fd24dd6716 : FAILURE in 10m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/689da479e9e9433bb2748d54ac67d428 : FAILURE in 14m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb70930a03c7496680762e162e30401c : SUCCESS in 18m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/534c0f93a98844c6a77b207af5cb894c : SUCCESS in 8m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/20d1029ca52a476a879882f536e13fdd : FAILURE in 13m 08s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/00becce238694096a7dd0eccf347cf34 : SUCCESS in 30m 06s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5d762f45cf0a459abeb61c7e09a790d4 : SUCCESS in 27m 00s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/d762c7d9df874e9fa8e5dd0fde5762f3 : SUCCESS in 1h 24m 21s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/33bbe5a6e40443908c09fc51464c6875 : SUCCESS in 49m 45s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e6fe43dfcc4947d2b35131fa40dd86b8 : SUCCESS in 53m 49s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c3680cfc339b49ae8a26db1ff1bcc60c : SUCCESS in 51m 56s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8914f692feed48feb1ff7c00d8525db9 : SUCCESS in 48m 16s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8500779466c8478aa81df918a3957243 : SUCCESS in 46m 10s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ba617b7605e54b819693a540e833e3ee : SUCCESS in 1h 05m 11s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/f0ddda75f704494b85f7bd4419404752 : POST_FAILURE in 51m 35s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ae38b16b920f4b7fa7b710dd6ad70262 : SUCCESS in 1h 47m 49s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/147fc4f090424afd96fa4e73e51307ac : SUCCESS in 1h 17m 32s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e6a721545cd143feb5d04ad69cee1d37 : SUCCESS in 42m 26s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/41d3449549634e3ca346cf1a0a993806 : SUCCESS in 49m 31s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ad235ddc44134829a1c90115a4e8b24e : SUCCESS in 1h 26m 28s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0e96af2a94604d5eb33f5821fb45fc74 : SUCCESS in 7m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/9a84d9158ce3499c932b49472b98afd1 : SUCCESS in 55m 56s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0cb38f05d03e426f8b5d8fe532ea3433 : SUCCESS in 1h 09m 18s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c6ffa193e78e43eaa43f9eeda1a558d6 : SUCCESS in 51m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/8f3e37978a34491f954efe3e5b18fd29 : SUCCESS in 58m 42s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/12f6c2df61994290a5f183e193ff359a : SUCCESS in 57m 27s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3828e9fdca74ab18de6d7e5b360c2e3 : SUCCESS in 26m 40s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/e513cd9ba86149d2911ce86564707031 : SUCCESS in 29m 05s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/4e911d65d7354f568a8a52ab74ac1128 : SUCCESS in 47m 59s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/b04409910c5a40fe922985f392cadd56 : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01708e9814294b40ced4235d8ff43e29a5f1ebfe","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-01 10:53:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1134 : SUCCESS in 3h 57m 43s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1389 : SUCCESS in 3h 14m 24s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1137 : SUCCESS in 3h 8m 53s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1129 : SUCCESS in 3h 8m 22s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1432 : SUCCESS in 3h 2m 2s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1424 : FAILURE in 2h 42m 32s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1137 : SUCCESS in 2h 6m 25s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1422 : SUCCESS in 2h 4m 3s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1143 : SUCCESS in 2h 0m 50s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1384 : SUCCESS in 1h 42m 57s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"1a2463d762f3cf453346953163c76939d94ed7a9","author":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"date":"2021-09-02 19:05:16.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"83e46f97bd408c92be960b2c76965ef881a07b9f","author":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"date":"2021-09-02 19:42:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bd4ef031b6c577e2a3974019b6499575aff7eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"date":"2021-09-20 18:43:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50f01a29c38a83a2132d9d33e740e80e5be7a2e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:44:03.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"6dead4cafd6b9b2ce618d7baf1544f98f12f8215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:44:06.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"82015b6b178a98d56326e16c7c90941b558c4eb9","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-20 18:44:24.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"d380b1fbedc187b9e0c14b34ae4a291ff7e6813f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-20 18:45:00.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"75b5743ef2d1775b8416d00223cf86e425787439","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-21 12:47:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1280 : SUCCESS in 3h 37m 47s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1285 : SUCCESS in 3h 21m 34s\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1288 : SUCCESS in 3h 4m 30s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1575 : SUCCESS in 2h 52m 8s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1294 : SUCCESS in 2h 32m 36s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1573 : SUCCESS in 2h 29m 25s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1288 : FAILURE in 2h 11m 44s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1583 : FAILURE in 2h 4m 59s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1535 : FAILURE in 1h 49m 23s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1540 : FAILURE in 1h 26m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"d8ec7d2089be8770a2887f7a8fa08c0c74e3fbdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"date":"2021-10-26 17:59:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e24024c9b4b4df8ab2a75560ada4cf01fc7ce0b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 18:11:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a7c44a2df0f4054bf63494a99d179ee : FAILURE in 8m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/367240200ee54ab89a4c32a6ebe45d66 : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27fb88034abc3c7c4aa1dbdcdf90840dbada149c","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-10-26 18:47:56.000000000","message":"Patch Set 3:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/51/806751/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/9654030/ : FAILURE in 18m 33s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":3},{"id":"456a13c68e4483fb3e76140ee7edf35e46aca76a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-10-26 18:56:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/7364915/7364915.html : FAILURE in 8m 58s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 : NODE_FAILURE in 0s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/156ae79/156ae79.html : FAILURE in 6m 52s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/1c823b7/1c823b7.html : FAILURE in 7m 41s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/7c9e249/7c9e249.html : FAILURE in 5m 40s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/51/806751/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/8190241/8190241.html : FAILURE in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23a36363c8fa8a36019dfd2ee52da69729e663a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:05:32.000000000","message":"Patch Set 3: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b87e644b588e42aab6b4c5ebc33ddf16 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a11f9245aea4b70a2f82b53088f24ef : FAILURE in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4ca5443efb84706bad0fa5d6cd3e692 : FAILURE in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbcc32b469d74eaaa19be594744d385f : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9991196a657d46938ad89e8679cf235e : FAILURE in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4d0ac54bec9436380a2c04988ccabc6 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2daced9a6b7f486e80aef7579ca5053d : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd520ddfbc14a998b3e1f1c01ec568e : FAILURE in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83864545984a49d5b369ebf51a5e9c66 : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fc9f8b22fbe345efb72f89ff450ee7c1 : FAILURE in 10m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/185a2f7915624278a9c66b4180c1aba5 : FAILURE in 22m 07s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/8e2ca633dcd04fff824b84455cc1c1ab : FAILURE in 13m 31s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e7781542fbd64fce9d6aeea9d647008d : FAILURE in 54m 29s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b64281e6d486443e81274fe8a3594ac1 : FAILURE in 22m 10s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6be0992d58e049978816fea463379eea : FAILURE in 26m 42s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/faddaedee2d74d42a02fdf63ed176fd5 : FAILURE in 18m 39s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b603a4d2fb0f4bc580d4e5c5daf28c2f : FAILURE in 29m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/6aa48a401e6e4742afa7334850a9ebfc : FAILURE in 26m 08s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b7f0f38e9f284cdc98f1d5cf21d99bfa : FAILURE in 21m 44s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d45afd9914d8405f8ba43cf0d6a9a420 : FAILURE in 30m 30s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/83321abdaaa147dfad3771129211fcbd : FAILURE in 34m 16s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/28b54c67a07b45bfb7d4cce64f538b4f : FAILURE in 26m 36s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/fab9c3d3f9364a8398be71bedf2776bc : FAILURE in 28m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ea016628f81c4ff2bf40401e8fda37e3 : FAILURE in 30m 29s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b77a6f17d5cd4f2981fb10c49b261494 : FAILURE in 20m 23s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/fe1924f9849d4253b955ed20eea66cf6 : SUCCESS in 4m 59s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c7feec1bc30342b9b6b135018dcbc5a9 : FAILURE in 29m 17s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/b1f5a3308ac7492c92999cb81a2fc7e4 : FAILURE in 30m 33s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8de9d36e12c646e19c01ad0fa0ab0ba4 : FAILURE in 24m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/e27ef94754684f6caedd5e3846eafc6f : FAILURE in 26m 06s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a1f8a34a4cf14416bfd6c51dde732b39 : FAILURE in 21m 14s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6985cd47a73045f5b948cf87535f3e46 : SUCCESS in 11m 50s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/124aeadb05b64b7dbaa6bf80829c06b3 : SUCCESS in 16m 53s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/36b94edb44184227a16ede164f338ae6 : FAILURE in 19m 29s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/7921a7d5570f42349daf2b364d2b3d92 : FAILURE in 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"041481ceb47988edded9a3921d5996187f4d4260","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-10-26 19:21:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/11e51895dfa54daf9e8812cebc4aff5b : FAILURE in 21m 33s\n- 14G-uefi-idrac https://100.82.32.82/t/openstack/build/31765e5c8283433da9c1b8f1539f86bd : FAILURE in 20m 54s\n- 14G-uefi-ipmi https://100.82.32.82/t/openstack/build/3eef550209f549708de7b0ee3e2846a5 : FAILURE in 21m 17s\n- XE2420-uefi-redfish https://100.82.32.82/t/openstack/build/73d5c17ff6ef4ec681305ce17f6386b8 : FAILURE in 21m 11s\n- R7515-uefi-redfish https://100.82.32.82/t/openstack/build/77a6c3fd5e894e21a3e4243bcdaff08d : FAILURE in 20m 43s\n- 13G-bios-idrac https://100.82.32.82/t/openstack/build/c51c9ec2f08e4464a244a086d534cdd0 : FAILURE in 21m 30s\n- 13G-bios-ipmi https://100.82.32.82/t/openstack/build/b8d0273c88b74bf2929a3fb6cc3a289b : FAILURE in 21m 09s\n- 14G-bios-idrac https://100.82.32.82/t/openstack/build/48d95caa8a204c5f970121b5ac0fa24d : FAILURE in 20m 59s\n- 14G-bios-redfish https://100.82.32.82/t/openstack/build/2090b244e1d842d7b0ac84efcc030d54 : FAILURE in 20m 40s\n- R6515-bios-idrac https://100.82.32.82/t/openstack/build/b39376a605044ff78a538540fede192a : FAILURE in 20m 53s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"59a8697db1a729bcfa4dbe0d0abd4c7e0e539924","revisions":{"35148bd7740141381e76233f5815b536ed103750":{"kind":"REWORK","_number":1,"created":"2021-08-31 13:47:45.000000000","uploader":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"ref":"refs/changes/51/806751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/51/806751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/51/806751/1"}}},"commit":{"parents":[{"commit":"3f17799a2e4da809ff7e4c956f5e5fcea0966cb0","subject":"Adds Attestation Interface to Node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3f17799a2e4da809ff7e4c956f5e5fcea0966cb0"}]}],"author":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-08-31 13:46:16.000000000","tz":-240},"committer":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-08-31 13:47:14.000000000","tz":-240},"subject":"Keylime attestation interface","message":"Keylime attestation interface\n\nAdds support for the Keylime attestation interface.\nDepends on https://review.opendev.org/c/openstack/ironic/+/755836\n\nChange-Id: I1b334fd820527415db21e301ee131adb67710b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/35148bd7740141381e76233f5815b536ed103750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/35148bd7740141381e76233f5815b536ed103750"}]},"branch":"refs/heads/master"},"5b39f4fbba04116d1d663b11bb39e284e5f2af06":{"kind":"REWORK","_number":2,"created":"2021-09-20 18:43:32.000000000","uploader":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"ref":"refs/changes/51/806751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/51/806751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/51/806751/2"}}},"commit":{"parents":[{"commit":"3f17799a2e4da809ff7e4c956f5e5fcea0966cb0","subject":"Adds Attestation Interface to Node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3f17799a2e4da809ff7e4c956f5e5fcea0966cb0"}]}],"author":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-08-31 13:46:16.000000000","tz":-240},"committer":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-09-20 18:41:51.000000000","tz":-240},"subject":"Keylime attestation interface","message":"Keylime attestation interface\n\nAdds support for the Keylime attestation interface.\nDepends on https://review.opendev.org/c/openstack/ironic/+/755836\n\nChange-Id: I1b334fd820527415db21e301ee131adb67710b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5b39f4fbba04116d1d663b11bb39e284e5f2af06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5b39f4fbba04116d1d663b11bb39e284e5f2af06"}]},"branch":"refs/heads/master"},"59a8697db1a729bcfa4dbe0d0abd4c7e0e539924":{"kind":"REWORK","_number":3,"created":"2021-10-26 17:59:13.000000000","uploader":{"_account_id":32454,"name":"Leo McGann","email":"lmcgann06@gmail.com","username":"ljmcgann"},"ref":"refs/changes/51/806751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/51/806751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/806751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/51/806751/3"}}},"commit":{"parents":[{"commit":"e3cf97f11a3ff2741519ce7222d9347bed3742e1","subject":"Adds Attestation Interface to Node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e3cf97f11a3ff2741519ce7222d9347bed3742e1"}]}],"author":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-10-26 17:58:14.000000000","tz":-240},"committer":{"name":"Leo McGann","email":"lmcgann06@gmail.com","date":"2021-10-26 17:58:14.000000000","tz":-240},"subject":"Keylime attestation interface","message":"Keylime attestation interface\n\nAdds support for the Keylime attestation interface.\nDepends on https://review.opendev.org/c/openstack/ironic/+/755836\n\nChange-Id: I1b334fd820527415db21e301ee131adb67710b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/59a8697db1a729bcfa4dbe0d0abd4c7e0e539924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/59a8697db1a729bcfa4dbe0d0abd4c7e0e539924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
