)]}'
{"id":"openstack%2Fironic~810381","triplet_id":"openstack%2Fironic~master~I1972b0a70173a259d72a8f9d1a2df4221038ac61","project":"openstack/ironic","branch":"master","topic":"add-yoga-python-jobtemplates","hashtags":[],"change_id":"I1972b0a70173a259d72a8f9d1a2df4221038ac61","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2021-09-22 10:11:16.000000000","updated":"2021-09-24 03:08:23.000000000","submitted":"2021-09-22 13:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810381-add-yoga-python-jobtemplates","meta_rev_id":"281c9e003b7471759bf98ffc8f0d7ed3c999caf7","_number":810381,"virtual_id_number":810381,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 13:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-09-22 10:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-22 10:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-09-22 10:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 10:11:50.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-09-22 10:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-22 10:56:25.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-09-22 11:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 08:55:22.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 03:08:23.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"c47322824c9d9b86bfd2e3d824775421191895ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-22 10:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8835eeffb9238a23ab1c5c46848289eb56579e7","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-09-22 10:11:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8300903ee8432042211d8f815fecfcfb35fa7c0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:39:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd2d673696934299906fc05317e29d84 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5dbc7b2c703c47daafb8c4cf46c90f33 : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"511da5d7b575f0c5ccb47081694a77b7d259d0aa","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-09-22 10:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6032ddc54e2ee93385a7ec5dd57bf6ca6427b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 11:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79be081b354b4776a2f5c27456343ccd : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe41f06d22564b15930c10de47640731 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20eca5f1084f40049562e2efb3ec7542 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c10d054344f6417d89ddbeffa42a8b72 : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b95160e2a4854ce596a59eaadce5aa4d : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11fb96b78cdd4cc09b14868f32e04b4a : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a764b9d28db41188aae38f4144500c8 : SUCCESS in 14m 19s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e091db0c973e43f2bea692c4e8343b10 : SUCCESS in 12m 15s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/73810a53854c46d1a6b6990d8ef83d08 : SUCCESS in 30m 10s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a54a537cd76f4178bd48f27e42aca837 : SUCCESS in 28m 39s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/00b326bbaf3b41f9a79aa7311347d70d : SUCCESS in 1h 19m 06s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/2afd36a10b5947ac905ef3316dbd1597 : SUCCESS in 51m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/04ac4edc4db54257a59cb4ff63535e1c : SUCCESS in 1h 02m 17s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9ada791ac19843579a0303eb79e929fa : SUCCESS in 51m 16s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f6e87133dac444fa9b7828e4ed20f222 : SUCCESS in 57m 42s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/199c986a3f054c7b823044f26ec4c094 : SUCCESS in 46m 28s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ffc97189b01f4f2baa7c9605ef839dae : SUCCESS in 1h 15m 41s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/9626f6a9722343b69314dafe6f38d0b4 : SUCCESS in 47m 28s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d3afc7685c0a4f268da83b7af423eb8f : FAILURE in 1h 35m 25s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/03bebbb8ecbe4d8e9cef486b1936ee09 : SUCCESS in 1h 23m 04s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/699eb9e59f1343488e4e475e3a03c794 : SUCCESS in 45m 35s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/564615c6180c423cb9402954ca9be98c : SUCCESS in 54m 33s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/84deab1ddc8346ec9f4d80827eea78cc : SUCCESS in 1h 29m 56s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ef9b70290eb14240a12f02e5b989ae25 : SUCCESS in 6m 55s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6fa1f68ba1b54bbda582ab34917ee8c4 : FAILURE in 48m 39s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/aa76a02ec6ec4b76b7a8bb6f7ce0d6cd : FAILURE in 42m 55s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3bb90acb69324c488dbd4c39b9305890 : FAILURE in 46m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/01fb9cbea0f44cf989ebb648738ea463 : SUCCESS in 51m 19s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a2ef7f74554842baad3d96fc8bdaaa55 : FAILURE in 46m 18s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0778ca951fa84dc59654b378de4ebe9e : SUCCESS in 26m 25s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/917f879ffe8740f6af47a7da82fce3d6 : SUCCESS in 29m 31s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/0882ea1507cf4eaeab1a69806dae6721 : SUCCESS in 59m 28s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/2ad47f771e874e96ad9db5ceeae9268c : SUCCESS in 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"113bae6f444c2452caa6d3b642297fd06f2300c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 12:04:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e9d8bbfc6f4df65e90f6ccb429f10fcf7b80ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:52:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc56708167b14552ac8932769efcab88 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d70e0318154778aad157a055b7fd50 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17b58ecad16846e2b930f054a11a03a6 : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1b9321918344a5da29190dd8ae8410f : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fc6b7fef304b2aa1711e353416d9bf : SUCCESS in 14m 28s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/92b4f6e6164d41d1a063c9b967510ff3 : SUCCESS in 14m 27s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c0a9b9a9c97c41cdad5fa3996f26f555 : SUCCESS in 31m 27s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/ed2917f64d6342a1b8573bce201768b9 : SUCCESS in 32m 03s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/07ff6beb57c746e2bca4718f89b1493d : SUCCESS in 1h 21m 05s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c14646aaeea0479084d4a038c3d73f75 : SUCCESS in 39m 30s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/6ee7b9b7af0549c585231d840bf2cdf3 : SUCCESS in 1h 13m 00s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2bc46be6f431497082f5f93651c7590e : SUCCESS in 51m 48s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4cab2b7fd9f04d9ea2ed392cb7d9c175 : SUCCESS in 57m 50s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/426abd8e661e457f94ceec9b8a63c0b1 : SUCCESS in 52m 01s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/162bde09bd9a443d8069787e59c02150 : SUCCESS in 1h 11m 36s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/bf2dbd8b69554964ab2d3252f77d0ad4 : SUCCESS in 54m 56s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c24fd260c16445f4afd27bcec735798a : SUCCESS in 1h 23m 40s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/5e8565d21e4b459caf97bf384422f79d : SUCCESS in 36m 49s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/65be82487097488698aa2a6cdebdba8b : SUCCESS in 56m 26s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/aef8caad632a4d61998a818e043bfd16 : SUCCESS in 1h 40m 02s","accounts_in_message":[],"_revision_number":1},{"id":"519ae8590836a72482db931a7465c535dd801565","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:52:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2163b7b40401899c871750a836c50e795a6a1515","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:34:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521131f31da14e1eb81166e1e573c797 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3732585c395e57776b7e9e4fe658398a93d68fa2","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-23 08:55:22.000000000","message":"Patch Set 1:\n\nTempest test passed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/81/810381/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/b9d18d9/ : SUCCESS in 1h 25m 31s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"281c9e003b7471759bf98ffc8f0d7ed3c999caf7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-24 03:08:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 14G_bios_idrac https://stash.dellemc-community.org/logs/14G_bios_idrac/1317 : SUCCESS in 4h 32m 37s\n- 13G_bios_ipmi https://stash.dellemc-community.org/logs/13G_bios_ipmi/1317 : SUCCESS in 4h 10m 5s\n- XE2420_uefi_redfish https://stash.dellemc-community.org/logs/XE2420_uefi_redfish/1604 : SUCCESS in 4h 6m 58s\n- 13G_bios_idrac https://stash.dellemc-community.org/logs/13G_bios_idrac/1323 : SUCCESS in 3h 52m 0s\n- 14G_bios_redfish https://stash.dellemc-community.org/logs/14G_bios_redfish/1314 : SUCCESS in 3h 34m 52s\n- R6515_bios_idrac https://stash.dellemc-community.org/logs/R6515_bios_idrac/1309 : SUCCESS in 3h 20m 46s\n- 14G_uefi_ipmi https://stash.dellemc-community.org/logs/14G_uefi_ipmi/1564 : SUCCESS in 3h 7m 43s\n- 14G_uefi_redfish https://stash.dellemc-community.org/logs/14G_uefi_redfish/1612 : SUCCESS in 3h 1m 27s\n- R7515_uefi_redfish https://stash.dellemc-community.org/logs/R7515_uefi_redfish/1602 : SUCCESS in 2h 46m 36s\n- 14G_uefi_idrac https://stash.dellemc-community.org/logs/14G_uefi_idrac/1569 : SUCCESS in 2h 32m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"816dbfdadaf97e391186b51b3de956ec18a22d18","revisions":{"816dbfdadaf97e391186b51b3de956ec18a22d18":{"kind":"REWORK","_number":1,"created":"2021-09-22 10:11:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/81/810381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/810381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/810381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/810381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/810381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/810381/1"}}},"commit":{"parents":[{"commit":"016c1f86074077459c7b410902c8dcb4ded39b53","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/016c1f86074077459c7b410902c8dcb4ded39b53"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-22 10:11:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-22 10:11:14.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I1972b0a70173a259d72a8f9d1a2df4221038ac61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/816dbfdadaf97e391186b51b3de956ec18a22d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/816dbfdadaf97e391186b51b3de956ec18a22d18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
