)]}'
{"id":"openstack%2Fironic~810487","triplet_id":"openstack%2Fironic~stable%2Fwallaby~I5018bf8ef6c86c6d303258f1497fa83d33b3cb76","project":"openstack/ironic","branch":"stable/wallaby","topic":"idrac-wsman-completed-with-errors-stable/wallaby","hashtags":[],"change_id":"I5018bf8ef6c86c6d303258f1497fa83d33b3cb76","subject":"Fix idrac-wsman having Completed with Errors jobs","status":"MERGED","created":"2021-09-23 06:36:42.000000000","updated":"2021-09-25 03:57:45.000000000","submitted":"2021-09-24 09:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810487-idrac-wsman-completed-with-errors-stable/wallaby","cherry_pick_of_change":809747,"cherry_pick_of_patch_set":1,"meta_rev_id":"a9894184a1facee16d3cc717adc3fd361f51113d","_number":810487,"virtual_id_number":810487,"owner":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 09:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-09-23 11:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-24 07:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-24 07:47:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 07:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 07:06:50.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-09-23 09:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 11:11:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-09-24 05:54:29.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 07:47:29.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-09-25 03:57:45.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"f443b6bdb56f877931e222460669f872c0dcd0d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-09-23 06:36:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b95c6c8390eca47dc910b9a387894d35d49a618e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 07:03:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df713eadf7a5481791e70c011773aa08 : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85699043d3be4ebbb07fafe047efb3fe : SUCCESS in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"696e6b7d66afa0042455758d004289bb0ebf2bc8","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-09-23 07:06:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/87/810487/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/af2544d/af2544d.html : FAILURE in 7m 35s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/87/810487/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/2051ba2/2051ba2.html : FAILURE in 7m 27s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/87/810487/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/c9cdb1d/c9cdb1d.html : FAILURE in 4m 33s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/87/810487/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/26d6244/26d6244.html : FAILURE in 5m 27s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/87/810487/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/e417946/e417946.html : FAILURE in 5m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 : MERGER_FAILURE (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d37e2ce09c99c09aba3b154af1fc8daf217b09fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:03:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a33da81f7d90496792982eb32f320cc2 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead5ce8cc31f4c7facee3817eb250ca4 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6effd42b6204425f88cd73fab0dcab79 : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cd21cd6469d46b5be8b22ef80c2bf17 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/303cbf757bef4cdb8f6e3019da01d842 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea72d824bb5f434eaa5a0ea8d2257039 : SUCCESS in 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a694f182c6104ba38c740032ec454cb2 : SUCCESS in 9m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/73241d399d904f82a20d43e36cbbb424 : SUCCESS in 11m 21s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/fccb6e5190f544f59440046f03be713d : SUCCESS in 28m 12s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/30999bf7a4ff4f36ab1d20877e63e9e4 : SUCCESS in 1h 16m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9a51e9d33dcf4852a2ca1f441d64f83e : SUCCESS in 57m 23s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a893ed0bb3214b4498ba9d54f185e89f : SUCCESS in 1h 14m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/00bcaf58af144e75b09d44f0679f8bc8 : SUCCESS in 56m 50s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/dd6139bd02224b42b2fd7906bc92029f : SUCCESS in 56m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/11d54166281247e199784ab5bd829117 : SUCCESS in 59m 30s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b877f56055bd4f5da89fe9497baae6a4 : FAILURE in 1h 13m 09s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/329439cbca864df09b3e237adc6566ba : SUCCESS in 55m 32s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7469b01186344ab1a9c983488300909c : SUCCESS in 2h 08m 11s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c12e1342225e484a8c11d484bfc6644f : SUCCESS in 1h 18m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/79310686e5d5406b92c6d61d1a001002 : SUCCESS in 43m 16s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a790392a6cc74fa48de7456bca8011d7 : SUCCESS in 52m 29s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e1f3cde5c7534551ac248293c13ce202 : SUCCESS in 1h 42m 39s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/cec52a74add2494b9e195ba53d23e833 : SUCCESS in 8m 13s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ff33cd083365493cae567ffdaf922ffa : SUCCESS in 1h 00m 32s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/f74b48674e4242d3975ae55d947b2fc1 : SUCCESS in 59m 45s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7d3748cf7a64453bb3e473f3944cc724 : SUCCESS in 57m 33s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/72f5ed33eb6843a1bdad097f2c2b6c54 : SUCCESS in 1h 03m 09s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a03ec19d526645ed8d2f6d82c1c92fb7 : SUCCESS in 27m 26s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/437e9d15975b47f38f206395f3d389c1 : SUCCESS in 32m 51s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/2a23e1deaf2648ee8def306219e82ec0 : SUCCESS in 59m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac6c23bc5bbcfdbe17cc3644606ea1dbb9bd9206","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-09-23 09:38:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d88bd7b9582aa51d0ad38c12866159494051225","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:57:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67166bf5581646d0bfd17e19ce51ecba : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32ca0d2cefe9461692ca94be4c6104d6 : SUCCESS in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"476b077ed543a5f827aa580c8b964868c369d0db","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-09-23 11:11:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3a062cb0075a1548e07ea0409ba1e0234683740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 11:12:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2113b61febfb4965a39aea0168bfc585 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f199fd526a4d24836743d539c5f0f4 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dbed58c35344f3b8a0e48c2399aa6ab : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a05e8bcfee64fe99b8b4a5dab314070 : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfcb910284834b11968376ef4cd69062 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a818128418944614957aa4cf05e1239a : SUCCESS in 16m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93a09c5e33c42edb1efe58934ae1975 : SUCCESS in 7m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/79295c69c1394c3faf7cf85a7e4796ff : SUCCESS in 10m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/aac30d0d6905435b899956bd826940e3 : SUCCESS in 39m 11s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/22d6403b65ce408a953f4fc8d7a29c9f : SUCCESS in 1h 25m 03s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/113dc3223ce245de81202515875c2227 : SUCCESS in 1h 05m 58s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/95bcbd4aa4424002b029a40ade655cac : SUCCESS in 1h 26m 26s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0210d7eb26d8475db26104c1ca9d1205 : SUCCESS in 54m 13s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d675594a53484253a2a6f8ee234f77da : SUCCESS in 1h 03m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/9cb58e6f008649b7a9117626ae0aa8ec : SUCCESS in 55m 34s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/25aad9036b634f9c9f3a3a1fe4704de6 : SUCCESS in 1h 22m 34s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/69220ece6f27475c8cdf217d8820a92d : SUCCESS in 55m 14s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/48ada7ba4d9c49ff96d364f0fad98b1d : SUCCESS in 1h 32m 04s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4fed3895e8f34be2b339eba1327241e6 : SUCCESS in 1h 24m 36s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c645dd7f539947f9bca3cdceee7532f8 : SUCCESS in 44m 06s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/05fd5f37aaa940838fba65ebb08ed90c : SUCCESS in 57m 59s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e8341313081842a0902b82b7a616b664 : SUCCESS in 1h 29m 37s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b699e26d6893487982d595f88463561c : SUCCESS in 4m 58s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/83e8613d86cc49708753e81c61accc0b : SUCCESS in 55m 43s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/45c7d414746148bb9fa1ca6ba3b9c90b : SUCCESS in 57m 40s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bababd9c792a46c0927614f5ebc98729 : SUCCESS in 53m 38s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/74e0a965644744f2a8903527de4d2d93 : SUCCESS in 50m 16s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6adf56a4c5524fa8849196dc7a79c99a : SUCCESS in 23m 16s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/b48d3a4165f94cd1b9426ceaf06f21bd : SUCCESS in 26m 29s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/56fb6b2525d2423f98eb4ad9218fe274 : SUCCESS in 53m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa5ff16ea0375e4ee5a00b8d29f5bef0feaa5342","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2021-09-24 05:54:29.000000000","message":"Patch Set 1:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nWe run following Tempest tests\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job http://openstackci.jp.fujitsu.com/irmcci/87/810487/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job/c449060/ : FAILURE in 50m 35s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f183481db902f39e623edd9bad79c9f471a0ed4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-09-24 07:47:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75dc2373d0ef6af0ec86355f917609ca68c36e2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 07:48:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"823be9441bf1c411fc8ef24edcc0ce44b492a22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:32:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931db05e4e064e88a0c1b5933bdc99ec : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77763e8a4b864ca6bcc63a50a6e946fa : SUCCESS in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daf32cf5b613456296c7c94b4adc6cbe : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f9cdbe499748df8afa842f141a42e9 : SUCCESS in 17m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c15a702ab464e229bf6fdda6c415018 : SUCCESS in 8m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/dd76f9276d8c43da83f120dbac35bcb0 : SUCCESS in 11m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2351c63d5ba84853a9f4b57e9130fb2f : SUCCESS in 31m 31s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/7e4234fb42404a2197b3346dd221fb24 : SUCCESS in 1h 18m 41s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bf8e984741084f46b8bb6ccfe98092f0 : SUCCESS in 1h 33m 52s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/3f1862fa2b004e5d93f0c911eb483e27 : SUCCESS in 1h 11m 30s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6bfe2b7d10cf44d085284a2d1d06e80f : SUCCESS in 51m 12s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/eebe6c8f619549baac004b232d68b08a : SUCCESS in 1h 06m 23s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f90d3ca58967459a8ce1c86f103326b0 : SUCCESS in 56m 12s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/bff162af0c0d44818e3d86be1a5ab331 : SUCCESS in 1h 22m 26s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/bdf70dc4da484cd28cd0e29d41084db0 : SUCCESS in 53m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ab0262d0286d42e8914319641eddf3d9 : SUCCESS in 1h 18m 08s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1060fdacd0cd42c8a94b3a507d05bb1f : SUCCESS in 43m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9432375a37c441e88f9f5dc4389cb4a1 : SUCCESS in 57m 39s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/e7a1df119a8f41f3ad4cbb15dc3f0dec : SUCCESS in 1h 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c9e8347233be1c6f2e8d899f8b995a7c35c3c742","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:32:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eaa2953f5e0d75021da08099e8032dee8e1a6d06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:33:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e8e83d672b04a5da81e4b2ace5d6b05 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c527f9d9dc034e3e9217e6ede130b591 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a9894184a1facee16d3cc717adc3fd361f51113d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-09-25 03:57:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 13G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/13G_bios_idrac_wallaby/82 : SUCCESS in 4h 37m 26s\n- 14G_bios_idrac_wallaby https://stash.dellemc-community.org/logs/14G_bios_idrac_wallaby/82 : SUCCESS in 4h 35m 10s\n- 14G_bios_redfish_wallaby https://stash.dellemc-community.org/logs/14G_bios_redfish_wallaby/82 : SUCCESS in 4h 1m 12s\n- 14G_uefi_ipmi_wallaby https://stash.dellemc-community.org/logs/14G_uefi_ipmi_wallaby/82 : SUCCESS in 3h 33m 46s\n- 13G_bios_ipmi_wallaby https://stash.dellemc-community.org/logs/13G_bios_ipmi_wallaby/82 : SUCCESS in 3h 33m 36s\n- 14G_uefi_idrac_wallaby https://stash.dellemc-community.org/logs/14G_uefi_idrac_wallaby/82 : SUCCESS in 3h 28m 2s\n- R7515_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/R7515_uefi_redfish_wallaby/82 : SUCCESS in 3h 22m 1s\n- 14G_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/14G_uefi_redfish_wallaby/82 : SUCCESS in 2h 54m 42s\n- R6515_bios_idrac_wallaby https://stash.dellemc-community.org/logs/R6515_bios_idrac_wallaby/82 : SUCCESS in 2h 28m 57s\n- XE2420_uefi_redfish_wallaby https://stash.dellemc-community.org/logs/XE2420_uefi_redfish_wallaby/82 : SUCCESS in 2h 25m 9s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d88bcf42d2995835c47ee1172e5df9cd96b9788","revisions":{"8d88bcf42d2995835c47ee1172e5df9cd96b9788":{"kind":"REWORK","_number":1,"created":"2021-09-23 06:36:42.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/87/810487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/87/810487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/87/810487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/87/810487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/87/810487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/87/810487/1"}}},"commit":{"parents":[{"commit":"c5beaa95c9ef7fe2471c6cabc6d49c29bf1a762a","subject":"Remove images from the OS profiler docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5beaa95c9ef7fe2471c6cabc6d49c29bf1a762a"}]}],"author":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-09-17 14:17:25.000000000","tz":-240},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-09-23 06:36:42.000000000","tz":0},"subject":"Fix idrac-wsman having Completed with Errors jobs","message":"Fix idrac-wsman having Completed with Errors jobs\n\niDRAC jobs can finish in \u0027Completed\u0027, \u0027Failed\u0027 and also\n\u0027Completed with Errors\u0027 state. This fix adds handling of\n\u0027Completed with Errors\u0027 as finished failed job otherwise node\nstays in wait state as it does not consider such jobs\nas finished.\n\nChange-Id: I5018bf8ef6c86c6d303258f1497fa83d33b3cb76\n(cherry picked from commit 6e0c0e7fd025dbfdc4f2e6070ad5cdb3079c2cef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8d88bcf42d2995835c47ee1172e5df9cd96b9788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8d88bcf42d2995835c47ee1172e5df9cd96b9788"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
