)]}'
{"id":"openstack%2Fironic~824014","triplet_id":"openstack%2Fironic~bugfix%2F19.0~I8bef819dc3101f88f4e04ea986e1039e34d917ed","project":"openstack/ironic","branch":"bugfix/19.0","topic":"story/2009774-bugfix/19.0","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-01-12 13:17:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8bef819dc3101f88f4e04ea986e1039e34d917ed","subject":"Add dhcp options for each ip_version once","status":"MERGED","created":"2022-01-10 17:28:23.000000000","updated":"2022-01-12 13:21:03.000000000","submitted":"2022-01-12 13:17:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824014-story/2009774-bugfix/19.0","cherry_pick_of_change":823817,"cherry_pick_of_patch_set":3,"meta_rev_id":"96351aa697ab83fca90aec4acef8d139d540ce0f","_number":824014,"virtual_id_number":824014,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 13:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-01-11 14:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-12 11:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-12 11:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 17:30:29.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-01-10 17:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-10 19:32:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 23:33:42.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-01-11 14:43:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-01-12 11:38:38.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"f1214c64354c3684586912eafff808de76e9d8e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-10 17:28:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"83eba0daa908ca3ba7e47f7f8c0c9b894f36fb09","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-01-10 17:30:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"e8088d02487bc5c6fd1fe9dc67936b4881a06d65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:52:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/122bdbeeeaba4e7fa842b0545d6a33da : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ac0ff747cd24f2a9d7aa7059e9e2cea : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e1cb18660946c7bcfae87ed912addfba68d1d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 19:32:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3de7f2d7c73e44c59b0c024168ebfb1c : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ab93f1404e4eb3837e4454689ee2c1 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7911c1e281124175a41622beb7547569 : SUCCESS in 11m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6f186fae6cb4a68a0469352285461ef : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5b271dc2bb441985f247b43cf16f9c : SUCCESS in 16m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dffac5ee52b64dd598bdda6f6ab20574 : SUCCESS in 9m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2d7ba75db7994d3bbc94030829b47811 : SUCCESS in 12m 38s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d4de7e67262e46249e4a538aab796b06 : SUCCESS in 10m 28s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/881c808977c54aea9bcf46f1eb0e5cd6 : SUCCESS in 27m 59s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c8030e99100a4aceadfb38f3a9c9b56e : SUCCESS in 24m 45s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/13a4838fabb247508168745655226858 : SUCCESS in 49m 49s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e960c81cc6d44e2d9fb5adfff99366ff : SUCCESS in 53m 33s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fc8c9b3a826b403aac9c0d20c223b8d4 : SUCCESS in 52m 33s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/980c137640e94803b3555e32d34532c0 : SUCCESS in 58m 00s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ad27de0408b7492a8d6d07a51b54b17a : SUCCESS in 53m 16s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/569d7b5bae5f442b90e66a5202e1730c : SUCCESS in 1h 34m 25s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/dbeb285f112740b48a141fab4bc91ed3 : SUCCESS in 55m 21s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4969b9fc60eb444ba593ba4282b2538d : SUCCESS in 1h 57m 37s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/2e88357aae3c4b888fa1b3fd6b67c152 : SUCCESS in 1h 21m 08s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/dd6ff85a88134401939ade1094290ac2 : SUCCESS in 39m 05s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/838ae58305b649fe8e0913b1feee3be5 : SUCCESS in 55m 03s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/6ed89efafc894d0e9b4da7893757b3b7 : SUCCESS in 5m 16s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/eaf7f8db62034d7193aeeaa97c6136f6 : SUCCESS in 54m 21s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/66fc92fbbcda4b2bbd2c0f653e5dfe35 : SUCCESS in 52m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/abc588696f1a429981dbc4fa62602662 : SUCCESS in 1h 01m 09s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c32454ee18e64b52b3db90f701aeff85 : SUCCESS in 55m 04s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22c83a7fa3fe4fda8b7ec60a54c4276a : SUCCESS in 29m 41s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/2cafd7fbd5ba434f9753d0d04efd39eb : SUCCESS in 35m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/86cd923f16924e798a03abc347ef6dca : SUCCESS in 54m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad06ed91ec878ef8d352b7c8e6583507045a73bb","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-01-10 23:33:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/93e4ab7bdc264fe2aa2b8e8b617a1143 : SUCCESS in 44m 51s\n- 14G-uefi-idrac https://100.82.32.82/t/openstack/build/fee6783fdcd34548bedecd3cf54ed8e3 : SUCCESS in 48m 06s\n- 14G-uefi-ipmi https://100.82.32.82/t/openstack/build/4471d483959b4bf4bdb96f52aafc4b80 : SUCCESS in 46m 18s\n- XE2420-uefi-redfish https://100.82.32.82/t/openstack/build/f9ec270b03a44e9f8b7d09fec6b2adaa : SUCCESS in 47m 36s\n- R7515-uefi-redfish https://100.82.32.82/t/openstack/build/022c2a88fc544cffaf1d4d0bbfe382af : SUCCESS in 46m 33s\n- 13G-bios-idrac https://100.82.32.82/t/openstack/build/311fd9a0cdbe458c8f85604d4f9c98da : SUCCESS in 49m 38s\n- 13G-bios-ipmi https://100.82.32.82/t/openstack/build/75f0bacca419407bbe484f3636f951aa : SUCCESS in 45m 59s\n- 14G-bios-idrac https://100.82.32.82/t/openstack/build/ffba29c99e054e7f803f5e4c689c7813 : SUCCESS in 49m 53s\n- 14G-bios-redfish https://100.82.32.82/t/openstack/build/34d8176086dc46d9bf21292bcba9b754 : SUCCESS in 53m 35s\n- R6515-bios-idrac https://100.82.32.82/t/openstack/build/ca8511af3b9c45c4b164f0fa7c11fe78 : SUCCESS in 50m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb51862a130e5f7c599e0b2ad4538be955804ecd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-01-11 14:43:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e213d268fcb502f2e491369d72f73eca0040f4b6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-01-12 11:38:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"190df8014478c4257e5e994a392b60843f24a784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 11:40:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b06138cf007d3fa7ff27ce8e843a046fc37ae2ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:17:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29dec9cc33f8480187c9e6664d659890 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94730c9555d41fbaa907b8a6198cc2e : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e2a9a0b08784baa8a83db511ca8bff7 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3176681251cc41768dbcf2612d9dcc85 : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a844b8204d4142379c9d5bf6f486245c : SUCCESS in 8m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fc779fb1c9cb41efb22c2f65ce495756 : SUCCESS in 12m 10s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/95166845615f4c7bbf132169b4c55c8f : SUCCESS in 27m 00s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/fd6bd27b47884cb190880fb8ef674211 : SUCCESS in 25m 14s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7271983bc394473386cf54c025db56a2 : SUCCESS in 53m 06s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/49d1e65bc5f1460cbea12607c9d9ab76 : SUCCESS in 59m 33s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d4253b490b944c1986c076d4e7ab84d1 : SUCCESS in 52m 21s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/246cc3ad5347445593ba702ce933d193 : SUCCESS in 56m 58s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/2660bf6354bc4745acfabb67ec90f680 : SUCCESS in 49m 20s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e632840890dc4cceaffd06538d08f223 : SUCCESS in 1h 33m 24s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/e57e4ccf2c104dc19b09de6ce6f125c9 : SUCCESS in 55m 49s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e51aba6ff9fb420ab059bc08a9ac15f4 : SUCCESS in 1h 13m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3a94384680464a6384cce8ff4b25ab49 : SUCCESS in 45m 09s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/cede7768164943f2890c4023d573741d : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cfcf1726cc45460634ffa69feb7b0f05fd1ae711","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:17:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"96351aa697ab83fca90aec4acef8d139d540ce0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:21:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd7963ba9a6424ca98652062524d56c : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9b07958719940bc98391036e43fa579 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a4ba7d27d31f2c7a6554bdb2e9e3dd03e001f58","revisions":{"9a4ba7d27d31f2c7a6554bdb2e9e3dd03e001f58":{"kind":"REWORK","_number":1,"created":"2022-01-10 17:28:23.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/14/824014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/14/824014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/14/824014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/14/824014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/14/824014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/14/824014/1"}}},"commit":{"parents":[{"commit":"b3ea14f863fec2c309eafc3099bfa0aa72ee3324","subject":"Configure CI for bugfix/19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b3ea14f863fec2c309eafc3099bfa0aa72ee3324"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-07 15:03:11.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-10 17:28:23.000000000","tz":0},"subject":"Add dhcp options for each ip_version once","message":"Add dhcp options for each ip_version once\n\nCurrently duplicate DHCP options are added when a port have\nmultiple address in the same address familiy. The duplicate\nentries cause an internal server error in neutron.\n\nThis fixes the issue by only adding DHCP options for each\nIP-version.\n\nStory: 2009774\nTask: 44255\nChange-Id: I8bef819dc3101f88f4e04ea986e1039e34d917ed\n(cherry picked from commit c4721da2d899e1c51e716577e776c6ee14e2e35e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a4ba7d27d31f2c7a6554bdb2e9e3dd03e001f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a4ba7d27d31f2c7a6554bdb2e9e3dd03e001f58"}]},"branch":"refs/heads/bugfix/19.0"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
