)]}'
{"id":"openstack%2Fironic~827481","triplet_id":"openstack%2Fironic~master~I75605cda40f039ef65c7c39058cbcaa449509013","project":"openstack/ironic","branch":"master","topic":"tempest-opts","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-16 09:13:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I75605cda40f039ef65c7c39058cbcaa449509013","subject":"CI: set the tempest plugin\u0027s partition_netboot option","status":"ABANDONED","created":"2022-02-02 11:38:14.000000000","updated":"2022-02-16 09:13:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"065f6d7917dc0ede5b8bdd271a63de1349eb0a6d","_number":827481,"virtual_id_number":827481,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 11:39:34.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-02-02 12:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 13:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 13:56:16.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-02-15 13:08:29.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b904464d2f40fbbc69eddda3047b5681c912530d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-02 11:38:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dde37dff90eac6c62db8848d9edd92a527860359","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-02 11:39:34.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1f3ef462e924c3f9eb401cae1b94ce60b92f3e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-02 11:42:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c36985de68c7e2ab27586848c7d528d61ca6976a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-02 11:43:45.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"b1d884de040d3a47224fdc91f89ca74181308da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 12:02:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd7712bacd3d4762b400b8af5e21350a : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecf1bb7d11d348c792f7cc8ad073b214 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d136b64545b84f4f7bbeb32db243dd691bf58e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:40:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af6534fbc32c4b16a5cf0951992b6081 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b786fba2c83f4c72999f09ec1a73f3ec : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d6d8dcabca4956929c0b7a95009bd0 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e941b238f24da3a2065e04fd03e82e : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ad88b1b86544dfd8ea0d857557b9d16 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab17871f73f46de9a457e76cd8705fb : SUCCESS in 13m 34s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0b724ba111cb471488eb8d830b206508 : SUCCESS in 11m 48s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2cfa8ab57ff64533affe31fee4561164 : SUCCESS in 28m 07s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/2808d6b024bb4310ac3797780cff3f9c : SUCCESS in 28m 17s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5b3b0556479c4f4c98749624184fafee : SUCCESS in 1h 16m 02s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/41505ac5f246439d8edbe03e33a1f819 : SUCCESS in 38m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4d5c5d9c2c3f4291817b1cb91629c5f7 : SUCCESS in 52m 59s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/81cd4dd68b6e4e24b993a9105c4bd854 : SUCCESS in 49m 54s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4b16beecad2c4ab0829c81c043059fb8 : SUCCESS in 52m 25s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4c58d7fa2b5b43bd99d01e3e4243171f : SUCCESS in 49m 43s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3cc823efd2ae4b808c5fbd2439286fa4 : SUCCESS in 1h 11m 23s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ba44d20be68b4bad94966bdb023398b0 : SUCCESS in 50m 52s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2255a5ab271543c58d26a1857b4d0f7c : SUCCESS in 1h 55m 48s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b239cb0ed00c4a759c1e0ddc2b66b22e : SUCCESS in 1h 21m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/22e2a9c4ccb74c82be30922108bc29fd : SUCCESS in 40m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ebb7819b030c48fba3325840dd0b74f7 : SUCCESS in 59m 43s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/41b08948be3f47c5965042bf4b8e5be1 : SUCCESS in 1h 49m 09s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ee474ea40c304c9eb27af06e35c26a31 : SUCCESS in 5m 23s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6777f09ae75e4e9b8f6309ef5b4a1154 : SUCCESS in 59m 43s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/136200a0d1de4cbb9fe4da0c2e4760f5 : FAILURE in 53m 03s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f4850004da424b23b7a18337b47379b9 : FAILURE in 39m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/a200eb9c03264c7ca00965aabd7f17d0 : SUCCESS in 57m 21s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/50d9362ef82445f99f897dbb13d7445e : SUCCESS in 52m 38s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54ec8df980704e03b892c5ac2bd404ab : SUCCESS in 31m 54s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/db0feabbef004b7c84466fa4be1c8840 : SUCCESS in 32m 26s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/14fcb02fc35848a1ae19fb77e9d25cf3 : SUCCESS in 59m 11s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/8cd250b24c8b46988bd6110952cc2b78 : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1fdb2c03e1fd8d0198ad72a66882217bfc920ef","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-02 13:56:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/b7a8339e50104e6c999c3242040c4530 : FAILURE in 23m 03s\n- 14G-uefi-idrac https://100.82.32.82/t/openstack/build/f89c6138c77643adbed8385a9c56adc1 : FAILURE in 15m 31s\n- 14G-uefi-ipmi https://100.82.32.82/t/openstack/build/3e08c31781204682a9ef3c132b955a66 : SUCCESS in 1h 30m 28s\n- XE2420-uefi-redfish https://100.82.32.82/t/openstack/build/874f702b65034c46bf9525422da2e94e : SUCCESS in 1h 18m 06s\n- R7515-uefi-redfish https://100.82.32.82/t/openstack/build/bd82baf237cf4ce4b7298acf17340c23 : FAILURE in 1h 05m 07s\n- 13G-bios-idrac https://100.82.32.82/t/openstack/build/0047d8c788a24102bced479972727075 : SUCCESS in 1h 21m 45s\n- 13G-bios-ipmi https://100.82.32.82/t/openstack/build/574580d1b54b41fbbb2f3f3461c8c7fd : SUCCESS in 1h 30m 36s\n- 14G-bios-idrac https://100.82.32.82/t/openstack/build/36fb8cede19043629a4856bc220ec534 : SUCCESS in 1h 33m 26s\n- 14G-bios-redfish https://100.82.32.82/t/openstack/build/4f5d0eaf66c24cd09d264b13cdcc48ef : SUCCESS in 1h 06m 12s\n- R6515-bios-idrac https://100.82.32.82/t/openstack/build/bd26002ed1a548b0b91b0c2c9bf85792 : SUCCESS in 1h 34m 28s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"4b3ab76f8d516d1c84aaff6dd291fee864f133e7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-14 17:08:45.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3633b7d2695f267e5d64a10edc3e4429287c3ff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-15 09:37:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b530dec20724b61ba3b620b302587a25bd97b1e5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-15 09:38:03.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"28a55a6f2d12cb844f34e06537842803e2384e78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:01:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4859b5d7493d4df896ce2a8efa4ce294 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82001762536a410a95ba0547aff011fb : SUCCESS in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0cdeefdb183bb6eab33cf82ece06d5f0dc9735b4","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-15 10:53:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/d143f9e/d143f9e.html : FAILURE in 42m 54s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/d06e0a8/d06e0a8.html : SUCCESS in 40m 19s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/3e53e4a/3e53e4a.html : FAILURE in 29m 33s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/23d9451/23d9451.html : SUCCESS in 38m 19s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/5537cd8/5537cd8.html : SUCCESS in 41m 46s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/85fa1c1/85fa1c1.html : SUCCESS in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bfbdc3973b7d6c6864c6ac74dae13cbee3f16e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:08:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8829e119b91a44efaccd13fd0eb08fd0 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5fb98cacde45f28fe31a6bb5536b79 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43878523f7c04a7dba52b0d2dfa1857b : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec7325802674061be7550c502e9e11a : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/537c0617da6f4e22bda3802dbdfcd862 : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790abd012fb240e3a54ba5dd63acd340 : SUCCESS in 17m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/449a9bdd9f634c239f4b5a2696cc83ca : SUCCESS in 11m 36s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7b4766965a3342fbb4b9f448d4ff49ab : SUCCESS in 25m 44s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/efa97aa806aa4a5aa436b1829f23afca : SUCCESS in 28m 59s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/53e5baf81ca748d9b8d5827b39fab6ea : SUCCESS in 1h 09m 26s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b8bd2ba3ef46409a9a2f2fd80eb671a0 : FAILURE in 27m 00s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d3af760f12e54f86bea0789d9db1ffa9 : FAILURE in 28m 29s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1b3f70757a9e47c1bdf04adcd825f5c0 : SUCCESS in 52m 04s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/78300c11a47644c0ba03268da47e9017 : FAILURE in 32m 44s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d4a7fd84eeb14519942ad4a365093d22 : FAILURE in 27m 47s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2310c3b6cc4f4604a8934791bd782157 : SUCCESS in 1h 27m 51s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/056661d0103f4d5a8c1906f2af40bb5a : FAILURE in 31m 12s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c2aa89b2d3a2412d93d3da3142480020 : FAILURE in 36m 18s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/22479cb723f64058bd586095adfc366c : FAILURE in 29m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b5791bba20e5499984ae93fbc5626ce8 : FAILURE in 30m 50s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/722443f398464b689cc1f45740a12cf0 : FAILURE in 27m 04s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/fc17fb861b0540baacbadaae36c4882d : FAILURE in 25m 43s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/a2925f9336df46e8b7725e72cb45ef14 : SUCCESS in 6m 51s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a6fc0399745442eda1d802d215c9b5e0 : SUCCESS in 57m 35s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/268baa427ef54b1d8ab0ada8ecb3ae0d : SUCCESS in 1h 02m 23s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d63609b829664716bff21f2282678849 : SUCCESS in 48m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/703679b1811c4657963dd7920dd730a9 : FAILURE in 28m 27s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/040fc05d671640eeaf51d2992b31df4f : SUCCESS in 55m 28s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd0520523d654e5ea5690c9701caafe3 : SUCCESS in 29m 56s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/5c077a70f8bb4aa49e28138f57a42d10 : SUCCESS in 37m 54s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/6e1ba40d6e4844d3a243ba3fdb741399 : SUCCESS in 59m 50s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/ae29aad840c64d9c8ee7a0f1b9d3e6b9 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3c4338c7c7f7967d021374f86317aaa5d5b6821","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-02-15 13:08:29.000000000","message":"Patch Set 3:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with PXE boot \u0026 the other is with virtual media boot,\nthen we run following Tempest tests in both environments\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-pxe http://openstackci.jp.fujitsu.com/irmcci/81/827481/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-pxe/2e17113/ : FAILURE in 37m 30s\n- fujitsu-irmc-ci-job-virtual-media http://openstackci.jp.fujitsu.com/irmcci/81/827481/3/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-virtual-media/d645240/ : FAILURE in 38m 09s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":3},{"id":"3dbbe01186e3066b99a142ed2477a3a301a0114f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-15 14:13:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"734e4dd057d68c39e34e8addd0962153bd3a54c9","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-15 14:14:19.000000000","message":"Patch Set 4:\n\nThis change depends on a change with an invalid configuration.\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"7e63e89eddc6af0e32e0f0e229e56e42f7d22189","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-15 15:44:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/b1e2fcd/b1e2fcd.html : FAILURE in 42m 09s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/509c771/509c771.html : SUCCESS in 39m 18s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/5c49ba4/5c49ba4.html : FAILURE in 29m 08s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/f08bb1c/f08bb1c.html : SUCCESS in 36m 47s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/1671378/1671378.html : SUCCESS in 37m 35s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/81/827481/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/0bd4534/0bd4534.html : SUCCESS in 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b09a94586cc883a4c45214bd5b23f5b8b835fa43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:04:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75b8ada219784b1f864e8bc7f40539c9 : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d69a310a416f46e1bcb21fd202d164cb : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91fa442f1784486aa987d00e2204b2c7 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffa40783bad4f2abdeeeeee0a73509d : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cebd094c4747439a8dcb0196a4b0be4c : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdaa38e7af0b43f9a910f3f05b519ccd : SUCCESS in 13m 33s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/80ecbcaedb764512bf04b66079b17137 : SUCCESS in 12m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a4f1584a30ac4917aa653d5186b15731 : SUCCESS in 28m 39s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/f5e45b9ee8514df3953f9015a0e4fc59 : SUCCESS in 28m 36s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/4014bf4b722d4ca787b6955def1fa73d : SUCCESS in 1h 14m 28s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b19def2f41e54addac1ef6c124e51fa0 : SUCCESS in 1h 03m 21s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b3b5490998ed4ee1875ff72fd12eb981 : SUCCESS in 1h 00m 25s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/07abb64848b04bdfb008db1e0bd2135d : SUCCESS in 50m 21s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/62a697fb234c419081a19fdda22d07e5 : SUCCESS in 53m 36s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/366c37167d24447f92af6af59b9914eb : SUCCESS in 51m 36s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d5b9e47bed144938b181362f843cb5fd : SUCCESS in 1h 08m 41s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/94e81165128649db93c118e914a3d9e4 : SUCCESS in 53m 17s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7862d567f4244b94bc327be1459be5bf : SUCCESS in 1h 34m 25s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0e81d6552d2e44d59c6483edb80094d0 : SUCCESS in 1h 23m 49s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7d5da462df30406098410b5be0bb824f : SUCCESS in 40m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c112840f51d54d7e848085d978de52cc : SUCCESS in 52m 30s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ef507171bfaf467daa56306163828c0f : SUCCESS in 1h 42m 23s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/41100efb04bf4118b4dfa2f0b3c6dc37 : SUCCESS in 5m 47s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6405f3ffae29437ba5b3ebf0223bcc80 : SUCCESS in 58m 39s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/27c99ddb64ce49a6af5529ee15c3ac0c : SUCCESS in 57m 25s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/335f1cc224dd4648aaa68d1488672434 : SUCCESS in 57m 43s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/1b70e2d9c2ac4c6f9295baa51614c1aa : SUCCESS in 57m 31s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/8d75d8be26cd4d0a9bb32651bc991da8 : SUCCESS in 1h 02m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e5cc607bf149f1bcb0ba7105793f11 : SUCCESS in 28m 49s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/a82a6f58c8884809a70db468daf0feb9 : SUCCESS in 35m 52s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/eeba3363cd684accb2e6d7d995f418de : SUCCESS in 56m 03s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/5982c023de314a9892e1efd499129769 : SUCCESS in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8e7fc5cde78c1caf91b1405e3bc5c6ff4a22ddc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:12:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70252e1ccdbc4ec0bd6fd5d75f95bbd4 : FAILURE in 30m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fcfd33220d54203833c30867cc2a07f : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d438c115b579596c77187099ce8d676c3abc4cd","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-02-16 04:42:49.000000000","message":"Patch Set 4:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with PXE boot \u0026 the other is with virtual media boot,\nthen we run following Tempest tests in both environments\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-pxe http://openstackci.jp.fujitsu.com/irmcci/81/827481/4/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-pxe/eea63ea/ : SUCCESS in 1h 09m 32s\n- fujitsu-irmc-ci-job-virtual-media http://openstackci.jp.fujitsu.com/irmcci/81/827481/4/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-virtual-media/6cec0ff/ : FAILURE in 48m 44s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":4},{"id":"88cd1db8c779686e1b763a1ce696a099e43dfa0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-16 09:06:45.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"b231b11d333b6d20ec162d9e47acd9ea737b6f04","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-16 09:07:02.000000000","message":"Patch Set 5:\n\nThis change depends on a change with an invalid configuration.\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"065f6d7917dc0ede5b8bdd271a63de1349eb0a6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-16 09:13:56.000000000","message":"Abandoned\n\nMerged in https://review.opendev.org/c/openstack/ironic/+/827882","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d38c4d4e7e31c69ad570fea17505cec8c3f84887","revisions":{"8a898e3c3e531dcc00d070f1e11185726440cdda":{"kind":"REWORK","_number":1,"created":"2022-02-02 11:38:14.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/827481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/827481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/827481/1"}}},"commit":{"parents":[{"commit":"5f6d753b2c334e4f404818d5e94a145b60244736","subject":"Merge \"Inspector: better error message on DiscoveryFailure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f6d753b2c334e4f404818d5e94a145b60244736"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"subject":"CI: expose default_boot_option to the tempest plugin","message":"CI: expose default_boot_option to the tempest plugin\n\nWe need it to work around the inability of CirrOS partition images\nto do local boot.\n\nChange-Id: I75605cda40f039ef65c7c39058cbcaa449509013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8a898e3c3e531dcc00d070f1e11185726440cdda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/8a898e3c3e531dcc00d070f1e11185726440cdda"}]},"branch":"refs/heads/master"},"0d3afc525d9bcc45ba4420f779e93b670f6dcd36":{"kind":"REWORK","_number":2,"created":"2022-02-02 11:42:25.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/827481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/827481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/827481/2"}}},"commit":{"parents":[{"commit":"5f6d753b2c334e4f404818d5e94a145b60244736","subject":"Merge \"Inspector: better error message on DiscoveryFailure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5f6d753b2c334e4f404818d5e94a145b60244736"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:42:03.000000000","tz":60},"subject":"CI: set the tempest plugin\u0027s partition_netboot option","message":"CI: set the tempest plugin\u0027s partition_netboot option\n\nWe need it to work around the inability of CirrOS partition images\nto do local boot. This option already exists but is never set\n\nAlso set the new default_boot_option parameter.\n\nChange-Id: I75605cda40f039ef65c7c39058cbcaa449509013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d3afc525d9bcc45ba4420f779e93b670f6dcd36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d3afc525d9bcc45ba4420f779e93b670f6dcd36"}]},"branch":"refs/heads/master"},"9b7ac5d122c27e8b7c8f8140b1374cf7f0011569":{"kind":"REWORK","_number":3,"created":"2022-02-15 09:37:42.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/827481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/827481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/827481/3"}}},"commit":{"parents":[{"commit":"10f464c01a0335f12308ff814c996d0696a6f623","subject":"CI: use a custom cirros partition image instead of the default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/10f464c01a0335f12308ff814c996d0696a6f623"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-15 09:37:16.000000000","tz":60},"subject":"CI: set the tempest plugin\u0027s partition_netboot option","message":"CI: set the tempest plugin\u0027s partition_netboot option\n\nWe need it to work around the inability of CirrOS partition images\nto do legacy local boot. This option already exists but is never set.\n\nAlso set the new default_boot_option parameter.\n\nChange-Id: I75605cda40f039ef65c7c39058cbcaa449509013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b7ac5d122c27e8b7c8f8140b1374cf7f0011569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9b7ac5d122c27e8b7c8f8140b1374cf7f0011569"}]},"branch":"refs/heads/master"},"9331742500db5116adbd0fd06e8590e28973f5b7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-15 14:13:56.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/827481/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/827481/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/827481/4"}}},"commit":{"parents":[{"commit":"a14dd6d67b9807efb3bfb4084586c8ced36e5ece","subject":"CI: use a custom cirros partition image instead of the default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a14dd6d67b9807efb3bfb4084586c8ced36e5ece"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-15 14:13:47.000000000","tz":60},"subject":"CI: set the tempest plugin\u0027s partition_netboot option","message":"CI: set the tempest plugin\u0027s partition_netboot option\n\nWe need it to work around the inability of CirrOS partition images\nto do legacy local boot. This option already exists but is never set.\n\nAlso set the new default_boot_option parameter.\n\nChange-Id: I75605cda40f039ef65c7c39058cbcaa449509013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9331742500db5116adbd0fd06e8590e28973f5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9331742500db5116adbd0fd06e8590e28973f5b7"}]},"branch":"refs/heads/master"},"d38c4d4e7e31c69ad570fea17505cec8c3f84887":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-16 09:06:45.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/81/827481/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/81/827481/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/81/827481/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/81/827481/5"}}},"commit":{"parents":[{"commit":"436358f8fa8d9315fa5fbaab61a3cb60b8f15c81","subject":"CI: use a custom cirros partition image instead of the default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/436358f8fa8d9315fa5fbaab61a3cb60b8f15c81"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-02 11:35:58.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-16 09:06:13.000000000","tz":60},"subject":"CI: set the tempest plugin\u0027s partition_netboot option","message":"CI: set the tempest plugin\u0027s partition_netboot option\n\nWe need it to work around the inability of CirrOS partition images\nto do legacy local boot. This option already exists but is never set.\n\nAlso set the new default_boot_option parameter.\n\nChange-Id: I75605cda40f039ef65c7c39058cbcaa449509013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d38c4d4e7e31c69ad570fea17505cec8c3f84887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d38c4d4e7e31c69ad570fea17505cec8c3f84887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
