)]}'
{"id":"openstack%2Fironic~827713","triplet_id":"openstack%2Fironic~stable%2Fqueens~Ie5b6a018e79bc578dbf8767e35be02e9d92e1893","project":"openstack/ironic","branch":"stable/queens","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-03-10 15:13:49.000000000","reason":"Julia Kreger replied on the change"}},"hashtags":[],"change_id":"Ie5b6a018e79bc578dbf8767e35be02e9d92e1893","subject":"Remove legacy experimental jobs","status":"MERGED","created":"2022-02-03 16:28:06.000000000","updated":"2022-03-10 17:53:30.000000000","submitted":"2022-03-10 17:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827713","meta_rev_id":"8a978c21b81e635e91b8e70637d8340e9dd2d0f5","_number":827713,"virtual_id_number":827713,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 17:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2022-03-10 15:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2022-03-10 15:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 16:28:25.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-02-03 16:29:27.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-02-03 18:40:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 05:55:59.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0157c5d7f11c0a0a110b646686f6f1bcbf9118a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-02-03 16:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3440902ad835bff6a308fbf15154c0a36451d1f5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-03 16:28:25.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch stable/queens.  The error was:\n\n  Unknown projects: openstack/ironic-lib, openstack/python-ironicclient\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-ironic-dsvm-base\n      parent: legacy-dsvm-base\n      irrelevant-files:\n        - ^(test-|driver-)requirements.txt$\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^install-guide/.*$\n        - ^ironic/locale/.*$\n        - ^ironic/tests/.*$\n        - ^releasenotes/.*$\n        - ^setup.cfg$\n        - ^tools/.*$\n        - ^tox.ini$\n      # NOTE: When adding to \u0027required-projects\u0027 also need to add a corresponding\n      # \"export PROJECTS\u003d...\" line in all the playbooks/legacy/*/run.yaml files\n      required-projects:\n        - openstack/ironic\n        - openstack/ironic-lib\n        - openstack/ironic-python-agent\n        - name: openstack/ironic-tempest-plugin\n          override-checkout: 1.5.1\n        - openstack/python-ironicclient\n      post-run: playbooks/legacy/ironic-dsvm-base/post.yaml\n\n  in \"openstack/ironic/zuul.d/legacy-ironic-jobs.yaml@stable/queens\", line 1, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"246384410f5128c82e84270e0892a64bb58b1937","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-03 16:29:27.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"aa9ab3166e05af57672fe41974c627059fb4b2cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:40:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23516009735444829a92dd962c8d3ae6 : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9eb06a394b1f4f0cbb59cfdab000c390 : SUCCESS in 8m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe32678eb560439185c249fffeba418e : SUCCESS in 8m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7527291285e4eaebe43d1a3f575d1ae : SUCCESS in 10m 43s\n- ironic-dsvm-standalone https://zuul.opendev.org/t/openstack/build/02569eaa3bbf4df89d80b47e25a8dc54 : FAILURE in 59m 55s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/231ffd6106314cd1873c2b27e3a3e8d9 : SUCCESS in 1h 11m 13s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/5d98954bca3140e296593942d197377a : SUCCESS in 56m 12s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/b2db853b69a644faa663258983c7ec3c : SUCCESS in 51m 12s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b6ce70af91324141b64ac0a772bb5985 : SUCCESS in 56m 26s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/9b03ca86da514dce8d7ca1e2a5676dc2 : FAILURE in 1h 22m 24s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/30da437a3505493a937a5454d257aefe : SUCCESS in 50m 47s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/fb524f082dee41939a59d628647870d2 : SUCCESS in 46m 37s (non-voting)\n- ironic-tempest-dsvm-ironic-inspector https://zuul.opendev.org/t/openstack/build/c32c2c9f31824828877b756fdbc31646 : SUCCESS in 50m 13s (non-voting)\n- ironic-tempest-dsvm-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/98aeb00511b9448e9bec25fb265b619a : SUCCESS in 47m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfaf202167f24511b9e55e0219783805 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9db370d4a792f299fed4364f924aa9bd6318b395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-02-08 18:40:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7f375d0463989e87dbf4d04fbf23ece9b37a3b50","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-08 18:41:17.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch stable/queens.  The error was:\n\n  Unknown projects: openstack/ironic-lib, openstack/python-ironicclient\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-ironic-dsvm-base\n      parent: legacy-dsvm-base\n      irrelevant-files:\n        - ^(test-|driver-)requirements.txt$\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^install-guide/.*$\n        - ^ironic/locale/.*$\n        - ^ironic/tests/.*$\n        - ^releasenotes/.*$\n        - ^setup.cfg$\n        - ^tools/.*$\n        - ^tox.ini$\n      # NOTE: When adding to \u0027required-projects\u0027 also need to add a corresponding\n      # \"export PROJECTS\u003d...\" line in all the playbooks/legacy/*/run.yaml files\n      required-projects:\n        - openstack/ironic\n        - openstack/ironic-lib\n        - openstack/ironic-python-agent\n        - name: openstack/ironic-tempest-plugin\n          override-checkout: 1.5.1\n        - openstack/python-ironicclient\n      post-run: playbooks/legacy/ironic-dsvm-base/post.yaml\n\n  in \"openstack/ironic/zuul.d/legacy-ironic-jobs.yaml@stable/queens\", line 1, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ed4c1923af351c3330e32d11d4ae2d9ad32b7da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-02-08 19:21:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae47b685c2e58cfa2c077ed1343ddcfffa08b9c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:20:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221c674dd921435480cb43511427ce71 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/467925d32e68457bab9f61d657f036fc : SUCCESS in 8m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b44ee5d2dd84de8884fca2753beb575 : SUCCESS in 6m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bce159603aa34c388589ff69622279fa : SUCCESS in 8m 22s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/3fc6264408e4475c875a13b0a0e5c620 : SUCCESS in 30m 36s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/b42542bd1bf345ab943272bd430fe2c0 : SUCCESS in 40m 09s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/7d305c4619c94bbba93d8ba7d4e7db8e : SUCCESS in 43m 00s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/05fbb35c3c074a29ad8b6989db0f8a24 : SUCCESS in 57m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/41ee9d8002f74ca38ef8ea3effbea751 : FAILURE in 1h 23m 48s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5a31db2c9b0140fcbbd5c896b257e15c : SUCCESS in 53m 15s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/38f38e7f38674563847e5f4f105475cb : SUCCESS in 48m 52s (non-voting)\n- ironic-tempest-dsvm-ironic-inspector https://zuul.opendev.org/t/openstack/build/ff4bebbdffaa48a99320d43d557cc6c3 : SUCCESS in 1h 01m 58s (non-voting)\n- ironic-tempest-dsvm-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/769f2497ff6c4417ac28ccc7e4c31616 : SUCCESS in 54m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa7ecf17c3c44611b93ca0e616ecfd41 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":2},{"id":"cee0822cd82aea3f25b0e84800342eb2d799b32b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 20:58:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4de413f552af571f1332bec3afef168a9bab20e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 21:58:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e554087d71634976a8d968fd07c4dbee : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7f3fc3c774944168f8893f027a65cc1 : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4df37145ed194c1a9199b9b995b82ed7 : SUCCESS in 7m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/54076cfefe204923afe539a72f40822d : FAILURE in 8m 23s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/cb9955bd3e9e4ce3ad3dd2a7893c2e41 : SUCCESS in 42m 44s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/dce8be4ac5d1498a983984000ee6c5e4 : SUCCESS in 1h 00m 08s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/c66dcf458133428ba35560ffb339bc2a : SUCCESS in 35m 48s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e25de9378874089bf59bdcd6a3212fb : SUCCESS in 49m 17s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/84cdfc9ac1254f79941c8c4c255820a2 : SUCCESS in 47m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f84cf962203e4217a0ef85d691e33944 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6d7e4553e237cf1b44d26879c6ce3c618d044efb","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-02-09 05:55:59.000000000","message":"Patch Set 2:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with PXE boot \u0026 the other is with virtual media boot,\nthen we run following Tempest tests in both environments\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-pxe http://openstackci.jp.fujitsu.com/irmcci/13/827713/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-pxe/26c4385/ : FAILURE in 7m 53s\n- fujitsu-irmc-ci-job-virtual-media http://openstackci.jp.fujitsu.com/irmcci/13/827713/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-virtual-media/c8e2abc/ : FAILURE in 8m 48s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"73883698e584f4d29cf8a2732f178abfc913921a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-04 17:56:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8927d7fba3dd359bcd0ed4b8175e01b78ba126dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 19:28:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b565eaf82242b9a856053001b86026 : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73ad33295043438aba7636a367912075 : SUCCESS in 9m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f74e8e3a20e94ce19406d2b3b347caf3 : SUCCESS in 9m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/adf148578b894ef3868decd9bebf4b95 : FAILURE in 8m 35s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/107b660a91e743c4881f7a22fbde721f : SUCCESS in 38m 42s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/cf9e68ca0cb64d56a2ba1d4d51c96a6c : SUCCESS in 53m 27s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/11dd0fc4ba794e4688c57b43c016b525 : SUCCESS in 48m 47s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b91f80df5d6945c7a64a558b6f7f779d : SUCCESS in 50m 08s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2628964817044724853f004fbc0ea5da : FAILURE in 1h 28m 03s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/26418502bd3d49de97865396c22277e0 : SUCCESS in 40m 38s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/863b0c69d00f4da692694ce83de441b8 : SUCCESS in 50m 54s (non-voting)\n- ironic-tempest-dsvm-ironic-inspector https://zuul.opendev.org/t/openstack/build/fa2ed7d5189c4e068481480d623e8bf6 : SUCCESS in 57m 12s (non-voting)\n- ironic-tempest-dsvm-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b3ca6813bcc645b494101f70488f2e75 : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd639419bd484936b119f047d3155024 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7a4115bbc0dd00daf1dc3b7b80ec72551019fd8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-10 15:12:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eec154e76e171a91814168caffedd4369c8223cb","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-10 15:13:14.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch stable/queens.  The error was:\n\n  Unknown projects: openstack/ironic-lib, openstack/python-ironicclient\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-ironic-dsvm-base\n      parent: legacy-dsvm-base\n      irrelevant-files:\n        - ^(test-|driver-)requirements.txt$\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^install-guide/.*$\n        - ^ironic/locale/.*$\n        - ^ironic/tests/.*$\n        - ^releasenotes/.*$\n        - ^setup.cfg$\n        - ^tools/.*$\n        - ^tox.ini$\n      # NOTE: When adding to \u0027required-projects\u0027 also need to add a corresponding\n      # \"export PROJECTS\u003d...\" line in all the playbooks/legacy/*/run.yaml files\n      required-projects:\n        - openstack/ironic\n        - openstack/ironic-lib\n        - openstack/ironic-python-agent\n        - name: openstack/ironic-tempest-plugin\n          override-checkout: 1.5.1\n        - openstack/python-ironicclient\n      post-run: playbooks/legacy/ironic-dsvm-base/post.yaml\n\n  in \"openstack/ironic/zuul.d/legacy-ironic-jobs.yaml@stable/queens\", line 1, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"5c33e060d5c4bf063140674380e3f8b5c9238170","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-10 15:13:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"07a27f5ee72b19410b4f21d4142f4b814ca42008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:42:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a823b50b45ca495a8fed9c319e1493d8 : SUCCESS in 8m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb925174e6d24c8893ecb0fae9c93958 : SUCCESS in 8m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54b5b3c9a2b743898557717ef1cb0a8b : SUCCESS in 8m 57s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/c2653917524f4b7abf51c530de4b085a : SUCCESS in 43m 25s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/8d523b5fab6f4a9d8b4003297ce314ae : SUCCESS in 52m 05s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/f3005d47f88544fbad1b9a50b0ec8cb0 : SUCCESS in 47m 25s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cbfa9c13e2654550bc9b277afdae88de : SUCCESS in 50m 36s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/033a7a3d9e9b4d4bac47502d572abbfb : FAILURE in 1h 17m 58s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6aa7bfc3b16c400a9a22ee6b76bc8358 : SUCCESS in 53m 02s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/a99b2d0deb2048aaa3fe667c7ec881ab : SUCCESS in 1h 01m 51s (non-voting)\n- ironic-tempest-dsvm-ironic-inspector https://zuul.opendev.org/t/openstack/build/00ef93c214534530907c832c4430651f : SUCCESS in 1h 12m 27s (non-voting)\n- ironic-tempest-dsvm-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/dce51805f1614440a245d5b225e457c4 : SUCCESS in 46m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01873c98e4674f30a8db252a1c0f11ec : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":3},{"id":"38e0986bfa104c59b3113a095423999f46e61fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:44:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea505e085f93863bcdef2ee37e523ccb3208e7b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:53:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f43e1a39d8492ca3da2c4c41c1f37f : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c25b4826a89408ba695522c8a241552 : SUCCESS in 8m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c812ae3796dc4e6dacd6a9a086694766 : SUCCESS in 9m 22s\n- ironic-tempest-dsvm-bfv https://zuul.opendev.org/t/openstack/build/c389561ad33b4d3ea739b26550733637 : SUCCESS in 47m 17s\n- ironic-tempest-dsvm-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/8013876f9f3643448b56473cd478416d : SUCCESS in 1h 02m 23s\n- ironic-tempest-dsvm-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/6ff10a81980f4028a15c575bffe2cf4e : SUCCESS in 52m 27s\n- ironic-tempest-dsvm-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1390472d62914154bce1600e351b2f33 : SUCCESS in 1h 01m 39s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5fdbdf9d3ae448f0b1d7aee026e77af4 : SUCCESS in 1h 03m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30181de9c12d4680ba09d80b453cb743 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":3},{"id":"8a978c21b81e635e91b8e70637d8340e9dd2d0f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:53:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7b5a86cf1673a19e9fdd66001fc6ec9fd7950cc1","revisions":{"ecac7554f9ba64cd840224ffbda6ff75707bca9e":{"kind":"REWORK","_number":1,"created":"2022-02-03 16:28:06.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/827713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/827713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/827713/1"}}},"commit":{"parents":[{"commit":"ad3cd1ba769864014d7e38947650e8fba893b467","subject":"Retries and timeout for IPA command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ad3cd1ba769864014d7e38947650e8fba893b467"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-02-03 16:26:44.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-02-03 16:26:44.000000000","tz":-480},"subject":"Remove legacy experimental jobs","message":"Remove legacy experimental jobs\n\nThis branch is no longer actively maintained and one of the\nexperimental jobs is invalid at this point in time. Removing\nit and a similar one since at this point it doesn\u0027t make sense\nto even keep the old experimental jobs around.\n\nChange-Id: Ie5b6a018e79bc578dbf8767e35be02e9d92e1893\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ecac7554f9ba64cd840224ffbda6ff75707bca9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ecac7554f9ba64cd840224ffbda6ff75707bca9e"}]},"branch":"refs/heads/stable/queens"},"09c06d83ba8328ee0fe27bfdea42a806abcb9ce4":{"kind":"REWORK","_number":2,"created":"2022-02-08 18:40:56.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/827713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/827713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/827713/2"}}},"commit":{"parents":[{"commit":"ad3cd1ba769864014d7e38947650e8fba893b467","subject":"Retries and timeout for IPA command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ad3cd1ba769864014d7e38947650e8fba893b467"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-02-03 16:26:44.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-02-08 18:40:56.000000000","tz":0},"subject":"Remove legacy experimental jobs","message":"Remove legacy experimental jobs\n\nThis branch is no longer actively maintained and one of the\nexperimental jobs is invalid at this point in time. Removing\nit and a similar one since at this point it doesn\u0027t make sense\nto even keep the old experimental jobs around.\n\nAlso, removed the standalone job. The standalone job by default attempts\nto run multiple standalone scenarios, which this branch is just not equipped\nto really support configuration wise.\n\nChange-Id: Ie5b6a018e79bc578dbf8767e35be02e9d92e1893\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/09c06d83ba8328ee0fe27bfdea42a806abcb9ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/09c06d83ba8328ee0fe27bfdea42a806abcb9ce4"}]},"branch":"refs/heads/stable/queens"},"7b5a86cf1673a19e9fdd66001fc6ec9fd7950cc1":{"kind":"REWORK","_number":3,"created":"2022-03-10 15:12:53.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/13/827713/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/827713/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/827713/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/827713/3"}}},"commit":{"parents":[{"commit":"ad3cd1ba769864014d7e38947650e8fba893b467","subject":"Retries and timeout for IPA command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ad3cd1ba769864014d7e38947650e8fba893b467"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-02-03 16:26:44.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-03-10 15:09:39.000000000","tz":-480},"subject":"Remove legacy experimental jobs","message":"Remove legacy experimental jobs\n\nThis branch is no longer actively maintained and one of the\nexperimental jobs is invalid at this point in time. Removing\nit and a similar one since at this point it doesn\u0027t make sense\nto even keep the old experimental jobs around.\n\nAdditionally:\n- Removed the standalone job. The standalone job by default\n  attempts to run multiple standalone scenarios, which this\n  branch is just not equipped to really support configuration\n  wise.\n- Removed the docs publishing job. At this point queens is\n  not going to receieve documentation updates, and the docs\n  jobs across all of openstack were impacted by sphinx\n  plugins going to python3 only, thus we\u0027re likely never\n  going to get doc publishing working again on this branch\n  without a community wide lift which is also impractical\n  given the current state of queens branch support.\n\nChange-Id: Ie5b6a018e79bc578dbf8767e35be02e9d92e1893\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b5a86cf1673a19e9fdd66001fc6ec9fd7950cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7b5a86cf1673a19e9fdd66001fc6ec9fd7950cc1"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
