)]}'
{"id":"openstack%2Fironic~830139","triplet_id":"openstack%2Fironic~master~I6cafa050d5c1c8289483f968d26c50485fd4528a","project":"openstack/ironic","branch":"master","topic":"gate","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-21 12:53:29.000000000","reason":"Removed by Dmitry Tantsur by clicking the attention icon"}},"hashtags":[],"change_id":"I6cafa050d5c1c8289483f968d26c50485fd4528a","subject":"CI: force config drive on the multinode job\u0027s subnode","status":"MERGED","created":"2022-02-21 10:58:25.000000000","updated":"2022-02-21 14:46:10.000000000","submitted":"2022-02-21 14:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830139-gate","meta_rev_id":"600386f174a4b8736f1c9ba7d9a68c943a1c922b","_number":830139,"virtual_id_number":830139,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 14:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-02-21 12:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-21 12:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-21 12:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 10:58:41.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-02-21 11:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 12:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 12:41:46.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 12:49:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-02-21 12:49:31.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"b9dbae52ccfd705ba4fa428060cadbbe0dd5d15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-21 10:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaa085078b8352c8ac34ca74136326843e26004d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-21 10:58:41.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch master.  The error was:\n\n  Unknown projects: opendev.org/openstack/ironic-inspector\n\nThe error appears in the following job stanza:\n\n  job:\n      name: ironic-inspector-tempest-partition-bios-redfish-vmedia\n      description: \"Inspect and deploy ironic node over Redfish virtual media using legacy BIOS boot mode\"\n      parent: ironic-tempest-partition-uefi-redfish-vmedia\n      required-projects:\n        - opendev.org/openstack/ironic-inspector\n      vars:\n        # NOTE(dtantsur): the inspector job includes booting an instance too.\n        # Excluding the abort tests since it hits \"node locked\" too often.\n        tempest_test_regex: test_baremetal_introspection\n        devstack_localrc:\n          IRONIC_BOOT_MODE: bios\n          IRONIC_DEFAULT_BOOT_OPTION: netboot\n          IRONIC_INSPECTOR_MANAGED_BOOT: True\n          IRONIC_INSPECTOR_NODE_NOT_FOUND_HOOK: \u0027\u0027\n          IRONIC_AUTOMATED_CLEAN_ENABLED: False\n          IRONIC_VM_COUNT: 1\n        devstack_plugins:\n          ironic-inspector: https://opendev.org/openstack/ironic-inspector\n        devstack_services:\n          ironic-inspector: True\n          ironic-inspector-dhcp: True\n\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@master\", line 264, column 3\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc99353533391b0ae669a8876f84a1166413926e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:23:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/858b422bf2c04685a8460c3d08ada5da : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abe0ba113cf5483488009aed1a61d8f5 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8467948d39277dadc27f2d6139f7bca591fb1b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 12:36:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ad5c3a4b55c4aea817173c0bbcb72f1 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/920d5dd4fc654005a715a54af74681a0 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b84d2f8a10141ebbbad4d60a408a90f : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c76a81a81514836bb6b0257cafd994a : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd77c2f6b9b24cd3bb377951cab275fe : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a0e1b1ee7b42c2a1f4322b13721661 : SUCCESS in 17m 29s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/879883d9ec494f5a97478329393f219c : SUCCESS in 12m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d0507dd6cae7468b89ef2f40ebb7ff71 : SUCCESS in 9m 46s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/51bfec08a66e498c83f2565e4beb140f : SUCCESS in 25m 02s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1b29f128a3ab4b98ace7feab77115791 : SUCCESS in 23m 56s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/9ba2d24efb10460a9c2fbd95570dfb80 : SUCCESS in 58m 54s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/dc783a675a4d484b9f95b0b6f3c208e7 : SUCCESS in 47m 21s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/79981bfc25024e169319dc15508241b9 : SUCCESS in 41m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1bbcdb5078144b1c82f961812d7fa7e4 : SUCCESS in 43m 51s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3a6a6ad9fb7847a7b1edb425f493f829 : SUCCESS in 49m 04s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f572bc57cfb94b99a260d9894a65d7db : SUCCESS in 43m 32s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/5d56ad6d8c444c51ae2c9ed858af0088 : SUCCESS in 1h 18m 33s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/0d8da9874c2d4037b90200e5d327fc22 : SUCCESS in 47m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1c02caaa05aa429aa2fed5241019436c : SUCCESS in 1h 34m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8d8dfdb7cb44477ca31324b3938db1ea : SUCCESS in 1h 04m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b9fd922116e54142b26fb1e382baa9dc : SUCCESS in 29m 50s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/756203df0aa04131a888957a9af2dcdd : SUCCESS in 41m 45s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/0b42894d8d51450bbefac112864dcf6d : SUCCESS in 1h 19m 11s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/989e07c4276844dbb5b69e08b2f4b948 : SUCCESS in 4m 43s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b8574977c76b41acad1cf9be598ebd8a : SUCCESS in 46m 01s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/8e12f9d11ed1480e982f1f5400b3231e : SUCCESS in 1h 02m 52s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c0a111eeeeab44d3b96669b2da7721af : SUCCESS in 43m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/34025d4ed361458bb444726e4b730a87 : SUCCESS in 56m 49s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/7b1b5aacbdb34a66b6d6db0d377ed265 : SUCCESS in 1h 10m 53s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6529b4b367a433fa471371edd691ed1 : SUCCESS in 24m 31s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/7f93a6eb92c74e9e992bd3d708c3cc4f : SUCCESS in 37m 07s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/257a4bed34e84f13884b12931339a0d7 : SUCCESS in 56m 42s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/e2f614ded9924c70b9846c0b9a2c194e : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"263a2f2b0a5cc063c88b2932681c048bc01cb040","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-02-21 12:41:46.000000000","message":"Patch Set 1:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with PXE boot \u0026 the other is with virtual media boot,\nthen we run following Tempest tests in both environments\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-pxe http://openstackci.jp.fujitsu.com/irmcci/39/830139/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-pxe/de68802/ : FAILURE in 52m 58s\n- fujitsu-irmc-ci-job-virtual-media http://openstackci.jp.fujitsu.com/irmcci/39/830139/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-virtual-media/fcfd99d/ : FAILURE in 42m 14s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"a4514e2cca3e7e6c3a209d6547919ec530ebf398","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-21 12:49:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22c8ec128b52899990d824156c8a0e0c362d6e35","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-02-21 12:49:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b927c90c02dbed358b1892a75edca79d135965f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 12:53:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dc036daf8fd444ab81d805786314de7a217cf12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 14:39:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b04d9c94b5524b8d94e38c1ac8d46df7 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feddd492d7ea423da0de746cc513110d : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f875635bd4646f985368bd8fd8d243c : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03ff313139944ff796b86c5119ccabef : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5da50ff0123433da044c78f7911636b : SUCCESS in 15m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/60f9df3195a4439abae81a4a8103a58f : SUCCESS in 12m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/4f1699bd8bd14827a4e3e1ac048f666d : SUCCESS in 32m 27s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4fa2bbcc61e8480786322aa7f18c1dc8 : SUCCESS in 32m 01s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0a88bf5f23a740e498866bd93aadbd95 : SUCCESS in 58m 29s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/5a5524ef949640708ae840a84aa783ba : SUCCESS in 44m 58s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/fadfc4d24eca442f93f41fd70a9aef97 : SUCCESS in 52m 52s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a363a0957cd44e4d892b0c7b92166679 : SUCCESS in 49m 57s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9af47d442c97434d85a003c19a9b75d8 : SUCCESS in 53m 39s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4833fb1b33ad42be84f56e2045160db1 : SUCCESS in 38m 19s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/82f330c877234b8c88ab7e83932fc61c : SUCCESS in 1h 26m 17s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/560944f99a0847a8a6c78bc1c7c8cd07 : SUCCESS in 54m 25s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/6dc7df2151744885916a5c20be9df6ea : SUCCESS in 52m 44s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/823d4fb872c4439582c8fe2b9a2835d9 : SUCCESS in 30m 54s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ad98cc9e48f54327876b531538548988 : SUCCESS in 48m 39s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/a51c3f0991524e03959653bde0d1dded : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":1},{"id":"67de16607a3800d99ada73934fd648adbfe50e89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 14:39:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"600386f174a4b8736f1c9ba7d9a68c943a1c922b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 14:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfcf6d792c4c42588c8e4db94147a2e7 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f09b7b1026551cb5818436a3714260fee9fcac4","revisions":{"2f09b7b1026551cb5818436a3714260fee9fcac4":{"kind":"REWORK","_number":1,"created":"2022-02-21 10:58:25.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/39/830139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/830139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/830139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/830139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/830139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/830139/1"}}},"commit":{"parents":[{"commit":"7ac480412626c38fa3493088dbf49e29303491b6","subject":"Build the new cirros image even when netboot is the default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7ac480412626c38fa3493088dbf49e29303491b6"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-21 10:57:31.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-21 10:57:31.000000000","tz":60},"subject":"CI: force config drive on the multinode job\u0027s subnode","message":"CI: force config drive on the multinode job\u0027s subnode\n\nWe need configdrives to pass information reliably, and the new cirros\nimage does not work without them.\n\nChange-Id: I6cafa050d5c1c8289483f968d26c50485fd4528a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2f09b7b1026551cb5818436a3714260fee9fcac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2f09b7b1026551cb5818436a3714260fee9fcac4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
