)]}'
{"id":"openstack%2Fironic~830355","triplet_id":"openstack%2Fironic~bugfix%2F19.0~Ia538907f3662e8cd84d988ea5d862c7f488558e1","project":"openstack/ironic","branch":"bugfix/19.0","topic":"gate-bugfix/19.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-22 16:31:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia538907f3662e8cd84d988ea5d862c7f488558e1","subject":"Build the new cirros image even when netboot is the default","status":"MERGED","created":"2022-02-22 10:58:31.000000000","updated":"2022-02-23 08:02:35.000000000","submitted":"2022-02-22 16:31:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830355-gate-bugfix/19.0","cherry_pick_of_change":829720,"cherry_pick_of_patch_set":1,"meta_rev_id":"cab14b5a85fc203aec2a188a58f64664dcf8ea6f","_number":830355,"virtual_id_number":830355,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 16:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-02-22 13:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-02-22 13:24:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 11:33:42.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-02-22 12:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 12:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 13:24:47.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-02-22 17:22:22.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-02-23 08:02:35.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"689d0924d57f031331a0a74087b6c40ce88bf193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-22 10:58:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed27cc91299c96190c598c33b8b3cc3566b4a912","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-22 11:33:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/6a9c1a8/6a9c1a8.html : FAILURE in 4m 27s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/cf37e82/cf37e82.html : FAILURE in 4m 38s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/3abe139/3abe139.html : FAILURE in 4m 38s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/9c4a13e/9c4a13e.html : FAILURE in 4m 33s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/8039099/8039099.html : FAILURE in 4m 28s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/55/830355/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/b3c7d9c/b3c7d9c.html : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b3b85c5dd84bfd1afbbb3b2193baab39ab37e2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:09:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5066304e240420aac0e50cb17af2140 : SUCCESS in 22m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6795a21310e4e969ca813ed2b584254 : FAILURE in 27m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4bad3e66521a3a06816be1ef6395f9ba6f26c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:56:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcd186b7f6204eaba0e658eccccb204d : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d5ac1f36f048a79306452dc58d01dc : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a639f786fc0e41da82282d0e176692bd : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f22312c7ba0f4ec79331b69c275b70fb : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c245bdbbc9cf425a9a5d3a64b53a5957 : SUCCESS in 10m 57s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/49de6bdf40234b6f8cbcbdf42e7d34cb : SUCCESS in 9m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/109b591038784024b28befade6f2649d : SUCCESS in 19m 10s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4d8457afcaa34a38a85ed9811f056f1b : SUCCESS in 24m 27s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/201f3e45b39645ac8d92869510038b47 : SUCCESS in 58m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c95e38c546414695b1341f0e55108c44 : SUCCESS in 58m 21s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/02ddccc512f64e4394159217e75517f2 : SUCCESS in 45m 24s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/41011e2fbd6c423ebee7f3ef6cb444ae : SUCCESS in 58m 27s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e02058d2ce9c47d09818039c9680cd16 : SUCCESS in 37m 12s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/72d89ef748904ce1839cac7849f6775f : SUCCESS in 1h 10m 56s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ebf5b133d68844dca49385984372f0a4 : FAILURE in 1h 09m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/88cca36db31d4da6b3cc0de6dfd7ee01 : SUCCESS in 1h 44m 13s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/682668c650a44d42af12db7ccb623cfc : SUCCESS in 1h 04m 07s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4cc5ba93ee044ac99fad65f01e8e9d3c : SUCCESS in 47m 38s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7dd3dc8920a6452180f12532d45b0b2c : SUCCESS in 53m 22s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/89f92b752d6d4e42aba7999c30930b2e : SUCCESS in 4m 36s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/bee8c3bbd3ec4e64aadfce80ca6c6961 : SUCCESS in 1h 02m 51s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fe11dd8ccc2b44f3858008c4ae7a55ee : SUCCESS in 1h 02m 25s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d39071a6a73f459285c77be9fc5b92b5 : SUCCESS in 45m 21s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/9242e4d59f594cfcac034157907f032b : SUCCESS in 58m 43s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b18882d2214e61a1e55418f4a8123f : SUCCESS in 31m 02s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/bd1304f9e1244ca186a608fed4366ddc : SUCCESS in 40m 07s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/2e1b5839ac7a44b4be69b54b5ee32417 : FAILURE in 1h 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1af69ce72d49014f5e77c4d41be2b23570b7abbc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-22 13:24:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57f1f5214e6de342b68a9bd6f89c0b452a1a94eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:58:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09030a2c6fd248f8850e093e969a8e26 : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/931cad66e3ba49cb86ea6f666144f5b2 : SUCCESS in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f34d3158f22458d596d30c42047aca1cdb1588c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:10:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2eb936d0acd48998426a491d0e6d14c : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b52f58206243a9ba793c715d8f17d6 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d30fc98c32d943b9a01d351028518572 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ca0042469964bc18cd185384bb23b84 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a42f4462584ea1b9aec9e18e3f654b : SUCCESS in 13m 23s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d893254b127646dd8cbc75b7b236677f : SUCCESS in 9m 58s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f519ecdf4bde4abf95615e45232f2302 : SUCCESS in 31m 03s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/f013ed12bbfc4dfb93f7c8131da9a847 : SUCCESS in 30m 25s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b68ed312c24c4f6b9b1aa0ec9f56e766 : SUCCESS in 1h 05m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b68712ad033c4b438cf3af1f0ce928b6 : SUCCESS in 1h 04m 41s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4d35013e4975417ea3d761bd4e14dd42 : SUCCESS in 53m 16s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3dc6142424c64e3680f2f3d86070a22c : SUCCESS in 56m 44s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/34ba58222deb49edb340ae715a61c68b : SUCCESS in 46m 04s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/8e541a3cc3d14bd9b8ed80ea69bcb510 : SUCCESS in 1h 11m 59s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/31263303482f42b2a00c807e84964b9b : SUCCESS in 57m 28s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3d3ba743402f49198ab185676b04e7b0 : SUCCESS in 1h 38m 54s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/74fdb866c9a946d888fe1dd1841bdf67 : SUCCESS in 1h 06m 41s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/cdf1688bef774019996242b014b523e0 : SUCCESS in 43m 28s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ec218966fe984ca6a4aba341c9af5638 : SUCCESS in 56m 54s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/639d928486c245458acd5d929ae978fc : SUCCESS in 5m 14s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/86db9753e6f5489ab9017cecb64fd47a : SUCCESS in 58m 37s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8f24d1465cc549449e1729bb4ac73e79 : SUCCESS in 56m 16s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/fef2ed58049e4882a2fb002e0bd976ef : SUCCESS in 48m 01s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/364300cc24a54a799c56d485c349cb13 : SUCCESS in 47m 11s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf0a5effd1f4a43a72dad74c1017969 : SUCCESS in 27m 33s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/67c412051e7c42b9a7d29bc1320545d3 : SUCCESS in 32m 47s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/83257eefb8c14d288f54be3c4d82988c : FAILURE in 1h 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"911c4ff6d5b909778bad9cb646bae7198b80ea58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:11:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d2e3c6d70c4d8abf5844bf44c160c0d4bf1f4ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:31:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1036bb3d7720436da8232d5bdc9f2d30 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c214ee41535848868550489f82ef8b0a : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d848058c004de1a666cadcdff7de98 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3651a6eac244ba98c04fa410c4b9e9 : SUCCESS in 14m 48s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5543cf1893d047a887b3c8ed0ad35b79 : SUCCESS in 14m 10s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e7a5353f138f4fc29827c4894aa42d59 : SUCCESS in 29m 44s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/734c6300da354a309f1fdbe63ea77587 : SUCCESS in 29m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d3c980f70ef94fdd878901d97793b563 : SUCCESS in 53m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/be0d27bf11bb4d65869e485c348ea8f1 : SUCCESS in 57m 20s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e6e7fe7e1d3f443eb79f2b8591e6e145 : SUCCESS in 54m 01s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/169f6542aa264787bd9ed4086041293a : SUCCESS in 56m 53s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/b8df9e5b0c9d4462b12a83e763986ba9 : SUCCESS in 49m 17s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/5dfaf9f6d019475da92cec7ace1b215a : SUCCESS in 1h 13m 18s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/edd8277f2055459fbbd1444fba65be01 : SUCCESS in 1h 05m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/39a143a49e994e9088c4f1aba349040d : SUCCESS in 55m 49s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/835654c975b742af897961ade32814e1 : SUCCESS in 42m 26s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3350321fd90442da85c188e4312950bc : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1b823c9042823cf98bd30f715871ce24b1e2757a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:31:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa00e5ae00fb770960f605175930ebfb251d06f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:34:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217d6310b97245d89df908065bb03302 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"404885a98b3a9123fc8b96f43eb897939644b4bf","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-22 17:22:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/b25897bdaa694410b0806b6e377b3491 : SUCCESS in 57m 09s\n- 14G-uefi-idrac https://100.82.32.82/t/openstack/build/ab44ca7cf5bd4194a3df120b52917de4 : SUCCESS in 1h 00m 32s\n- 14G-uefi-ipmi https://100.82.32.82/t/openstack/build/af358180820f4e7ab5a5283dcaeeda80 : SUCCESS in 58m 12s\n- XE2420-uefi-redfish https://100.82.32.82/t/openstack/build/40cdeadb120144bd9e9a22029e2e7b29 : SUCCESS in 56m 55s\n- R7515-uefi-redfish https://100.82.32.82/t/openstack/build/c50b6cc83f5c4d0f936e3cf1508afe6b : SUCCESS in 1h 01m 52s\n- 13G-bios-idrac https://100.82.32.82/t/openstack/build/027cf97bb13a4a59ad38eb39acd96126 : SUCCESS in 1h 01m 42s\n- 13G-bios-ipmi https://100.82.32.82/t/openstack/build/8b19b4ec81be48d1bd2a829ed4185665 : SUCCESS in 1h 02m 29s\n- 14G-bios-idrac https://100.82.32.82/t/openstack/build/0bfc9735594345a6934f69732382d3ae : SUCCESS in 52m 09s\n- 14G-bios-redfish https://100.82.32.82/t/openstack/build/a0bab47419ef4d71b51b52e1e2bd44d3 : SUCCESS in 50m 19s\n- R6515-bios-idrac https://100.82.32.82/t/openstack/build/e070ecbd96684cd189b4f4cd8b1a8874 : SUCCESS in 56m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"cab14b5a85fc203aec2a188a58f64664dcf8ea6f","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-02-23 08:02:35.000000000","message":"Patch Set 1:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with PXE boot \u0026 the other is with virtual media boot,\nthen we run following Tempest tests in both environments\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-pxe http://openstackci.jp.fujitsu.com/irmcci/55/830355/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-pxe/238ce02/ : SUCCESS in 1h 08m 21s\n- fujitsu-irmc-ci-job-virtual-media http://openstackci.jp.fujitsu.com/irmcci/55/830355/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-virtual-media/68debac/ : FAILURE in 56m 24s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bc81d7cc64b946acab0ecee1670b911e80b1f33","revisions":{"9bc81d7cc64b946acab0ecee1670b911e80b1f33":{"kind":"REWORK","_number":1,"created":"2022-02-22 10:58:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/55/830355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/55/830355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/830355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/830355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/55/830355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/55/830355/1"}}},"commit":{"parents":[{"commit":"cba1c2acc14d008a876989f8ef9bdff13048afea","subject":"CI: use a custom cirros partition image instead of the default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/cba1c2acc14d008a876989f8ef9bdff13048afea"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-17 14:36:11.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-22 10:58:31.000000000","tz":0},"subject":"Build the new cirros image even when netboot is the default","message":"Build the new cirros image even when netboot is the default\n\nThe standalone job changes boot_option in runtime, so local boot\ncan be used even when the default boot option is netboot.\n\nChange-Id: Ia538907f3662e8cd84d988ea5d862c7f488558e1\n(cherry picked from commit 7ac480412626c38fa3493088dbf49e29303491b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9bc81d7cc64b946acab0ecee1670b911e80b1f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9bc81d7cc64b946acab0ecee1670b911e80b1f33"}]},"branch":"refs/heads/bugfix/19.0"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
