)]}'
{"id":"openstack%2Fironic~833745","triplet_id":"openstack%2Fironic~stable%2Fwallaby~Id9428518d21290fb38a0f7471a2cb69a6cb3ffb2","project":"openstack/ironic","branch":"stable/wallaby","topic":"anaconda-rebuild-fix-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"6618":{"account":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"last_update":"2022-03-28 12:34:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9428518d21290fb38a0f7471a2cb69a6cb3ffb2","subject":"Fix rebuilds using anaconda deploy interface","status":"MERGED","created":"2022-03-15 13:43:29.000000000","updated":"2022-03-28 12:36:06.000000000","submitted":"2022-03-28 12:34:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833745-anaconda-rebuild-fix-stable/xena-stable/wallaby","cherry_pick_of_change":833558,"cherry_pick_of_patch_set":2,"meta_rev_id":"45830ce3aaef0432c51a0cb03e3c45c0d5d9b264","_number":833745,"virtual_id_number":833745,"owner":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 12:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2022-03-28 09:28:55.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2022-03-20 21:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2022-03-28 09:28:55.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 13:44:49.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-03-15 14:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-15 15:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 15:53:21.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-03-20 21:58:34.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2022-03-28 09:28:55.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"a2c5188480e85062c179c6f90ee355ef65b3e880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2022-03-15 13:43:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"990d1bc651bb6e35a11487371ec2b8ce507bc00b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-15 13:44:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"abed4b1b53a836344103d4d57d46f141ada97544","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 14:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cb7d53423cb487a8cc4e9bab40658a9 : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47e6a0e6f7cd4defa17d16972f198ac2 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee46f361b6bc3c0d00177c87c8253ffaa9376f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ef273c2044d49e6b169bd06167d3803 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6560eb502ad94b5e8ab1f1235c10ff7e : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3032efc75d14e5ebe47436b8a680e45 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b83dd9bb29f94b328d3939529287b1b6 : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90e3357793944fba91f52f369b3f627 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18c949881c14a95a14dec6fc7469482 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/855d9b842c6a4d22a4dfff5dd3ad318c : SUCCESS in 9m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/32d262e8d30742ff81854b1e3e02493b : SUCCESS in 13m 25s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b10c6aa3da8d48969e4b7d46c61a4b70 : SUCCESS in 27m 09s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/740dc143acef4a219488c5f488f1e8fb : SUCCESS in 1h 17m 25s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9fa6bb310b97422a805288f40dae426d : SUCCESS in 1h 07m 29s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/1d728622250c449abd9d3947aed9ae38 : SUCCESS in 1h 10m 20s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/db9eab731af442d681e6389b4aa5a671 : SUCCESS in 57m 06s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/cb60225bf6dd4bc78c65a5d36eb4f081 : SUCCESS in 1h 00m 49s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/265fc5a161f547249ddb49ce4f1da62f : SUCCESS in 51m 22s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/4c7030d814f249a1ac22244b7312998c : SUCCESS in 1h 11m 31s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/18bd6c79c7484bdebeac4155793cf77d : SUCCESS in 1h 01m 13s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ec6f521f0a1748549a793c581605ac17 : SUCCESS in 1h 57m 02s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0e97a50c8ea64bdbb4c509ed27ff1559 : SUCCESS in 1h 11m 14s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2d6e5fbfc09d4af892a3f61b110813d8 : SUCCESS in 44m 52s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/41a32dfbbe1948e4a70bd14f901617d8 : SUCCESS in 57m 59s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ab10dddc2a02476a976498b79cb0cd1d : SUCCESS in 1h 51m 13s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/92871f74a4b841a9b9fe4da21d088e08 : FAILURE in 6m 40s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/29b65c0a7db345048965b24b24a4d722 : SUCCESS in 1h 01m 35s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/6b2dd623f5cf4a37813929cb8f73e938 : SUCCESS in 1h 07m 30s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8b6a6095be88426cbb87dcac29381f28 : SUCCESS in 1h 00m 04s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f21b791908c44014b8b4dc6f7d608269 : SUCCESS in 56m 04s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16a3db20723d400e9ba4fa9575beea43 : SUCCESS in 26m 53s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/3075b0b806004246b2a313c8f63bd12e : SUCCESS in 29m 30s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/c378717aef7740e3b2aa70668051a38f : FAILURE in 1h 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b893d4256022a22a644ea13a2a38ba44079aea5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-15 15:53:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/5ecf726ee60c49e3b1f9679e144fa16c : SUCCESS in 1h 15m 54s\n- 14G-uefi-idrac https://100.82.32.82/t/openstack/build/74494432ec9e4228bdcf59caf4dfa5ef : SUCCESS in 1h 02m 11s\n- 14G-uefi-ipmi https://100.82.32.82/t/openstack/build/1751b43920d8496a91ec150bff56e758 : SUCCESS in 1h 03m 20s\n- XE2420-uefi-redfish https://100.82.32.82/t/openstack/build/3c5e6745cf8d46f887f32cf54237d7e6 : FAILURE in 1h 01m 17s\n- R7515-uefi-redfish https://100.82.32.82/t/openstack/build/bbe6b6d02e6d4db7b2541f70dcb25839 : SUCCESS in 55m 05s\n- 13G-bios-idrac https://100.82.32.82/t/openstack/build/424714aed1164d6ba9e0f6a1a0c25d3e : SUCCESS in 58m 24s\n- 13G-bios-ipmi https://100.82.32.82/t/openstack/build/8b6772e41178494c98d2826d2a7991dd : SUCCESS in 55m 37s\n- 14G-bios-idrac https://100.82.32.82/t/openstack/build/6bf041c917bc41cd9bacdda72a3c0d85 : SUCCESS in 53m 26s\n- 14G-bios-redfish https://100.82.32.82/t/openstack/build/45771052e8ca4d4da3b92f2f91dfdfd6 : SUCCESS in 54m 05s\n- R6515-bios-idrac https://100.82.32.82/t/openstack/build/ebcaa432938d490494e0a62f9aa59d28 : SUCCESS in 1h 06m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"fecfcb3d1c954372e2b51cf3170ad5a6df0770dd","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-03-20 21:58:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"144dd783cfa32c2f3730c7c3b7ec33eba6d3adc4","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2022-03-28 09:28:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb77adbc2d37e4fe45b6600f6162f73c447e5193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 09:30:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c389108fd1547405538d5a289b6dedc95f9fc6be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:34:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed89de3f2194f68ab2b17f5da8e7028 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d92f0075d94fa687cf7a18e8ce7f7c : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7797fedca39a4e868dd5e29b2a9705f9 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d09d7202214677868cd62a5e2e8470 : SUCCESS in 17m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c38a749361848748c0f964e23c76dcf : SUCCESS in 8m 45s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/85ca829716d24a75b28403b474774a17 : SUCCESS in 12m 31s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ce4bd98e0b684015b6eaac209dece5b8 : SUCCESS in 32m 41s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/e4f9a3f61fbd470898869cdac9a6919e : SUCCESS in 1h 14m 07s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/724c434715e74dbb8fd041bf45fd8550 : SUCCESS in 58m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/cebcef5775be48f2a219d58deae5b98d : SUCCESS in 1h 16m 04s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/36a192e98aa24b8cbd51a9323d52fb71 : SUCCESS in 59m 06s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4770aaae96804e098414323d4035e96a : SUCCESS in 1h 02m 01s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/502d8ae3d80740a1b6528bdb6d26b4f4 : SUCCESS in 50m 12s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/3c94acbc0c284c60b96d4f5106d1cf6e : SUCCESS in 1h 34m 16s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/4caa62131b15482fbc6cab2064f44f96 : SUCCESS in 59m 49s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b7959c7e1e3b4981aef7bf69f821d442 : SUCCESS in 1h 12m 31s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e814ae68ac75480099b4ccfdfea82bf9 : SUCCESS in 41m 36s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d0b5a0526a2e418abe358b2169e7c50e : SUCCESS in 53m 02s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/f621c2ed942942f88d5e8e3f70c318f9 : SUCCESS in 1h 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4dfd10602bb32804a7a7a83788c9d9d2e7c6f19f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:34:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"45830ce3aaef0432c51a0cb03e3c45c0d5d9b264","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:36:06.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed64024fce63477eaa677376efa1c353 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8933f45633084ddeb7d449a3dbbc5a94 : FAILURE in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22e0671aed4d6b61de687f721db041e15365a644","revisions":{"22e0671aed4d6b61de687f721db041e15365a644":{"kind":"REWORK","_number":1,"created":"2022-03-15 13:43:29.000000000","uploader":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"ref":"refs/changes/45/833745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/45/833745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/833745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/833745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/833745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/45/833745/1"}}},"commit":{"parents":[{"commit":"3b13211607d1ed954f4a9ae7bf34b388f81085f2","subject":"Merge \"Build the new cirros image even when netboot is the default\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3b13211607d1ed954f4a9ae7bf34b388f81085f2"}]}],"author":{"name":"Ruby Loo","email":"rloo@verizonmedia.com","date":"2022-03-04 15:27:56.000000000","tz":0},"committer":{"name":"Ruby Loo","email":"opensrloo@gmail.com","date":"2022-03-15 13:43:29.000000000","tz":0},"subject":"Fix rebuilds using anaconda deploy interface","message":"Fix rebuilds using anaconda deploy interface\n\nThe anaconda deploy interface was saving internal information in\nthe node\u0027s instance_info, in the user-facing \u0027stage2\u0027 and\n\u0027ks_template\u0027 fields. This broke rebuilds using a different image\nwith different stage2 or template specified in the image properties.\nThis has been fixed by saving the information in the node\u0027s\ndriver_internal_info instead.\n\nAddresses comments/nits from\nhttps://review.opendev.org/c/openstack/ironic/+/827924.\n\nConflict: ironic/common/pxe_utils.py \u0026 ironic/drivers/modules/pxe.py\nbecause Node.[set|del]_driver_internal_info() are only avail in\nmaster; changed to copy the dict, make the changes, and then set\nthe node\u0027s driver_internal_info to the updated dict.\n\nChange-Id: Id9428518d21290fb38a0f7471a2cb69a6cb3ffb2\n(cherry picked from commit ab68c9d88b76e4ff83f9dffddcc204676fdb50d5)\n(cherry picked from commit 42ec7df183591acf0078982e3385893585b86fbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/22e0671aed4d6b61de687f721db041e15365a644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/22e0671aed4d6b61de687f721db041e15365a644"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
