)]}'
{"id":"openstack%2Fironic~836699","triplet_id":"openstack%2Fironic~stable%2Ftrain~I996e5d1507363371860fe8d60dd047bc362dc8ff","project":"openstack/ironic","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"31616":{"account":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"last_update":"2022-04-13 20:13:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I996e5d1507363371860fe8d60dd047bc362dc8ff","subject":"[DNM] Testing Dell CI","status":"ABANDONED","created":"2022-04-05 22:00:30.000000000","updated":"2022-04-13 20:13:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e563be94241c1b969f47cb4d8a3d8e0deb038407","_number":836699,"virtual_id_number":836699,"owner":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-05 23:20:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 22:01:49.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-04-05 23:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 09:29:00.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"b94cb24ace05cbca0c3fba5b673cf02b2cfa2c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"date":"2022-04-05 22:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eabe7ddc0ec19a9b3dade38514adfa7bfd0652f7","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-04-05 22:01:49.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c015ab268c2d1460cd5c5a21dfa45d2c88e2831c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 23:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8e27537023540dabf12388919dfc5ed : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bcf0745457649799140fa3cba273f77 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c98d4845878494e8517949cc91a32dd : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41ac3671bab04c96b64037051c75f93e : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54555b9d0fec44409b47d78184dd29c1 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb582806e1049d1bc43fdc7a5ddde7b : SUCCESS in 15m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d30cdbedcec041b39af5c10724de040d : SUCCESS in 9m 37s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/a6d0effec81547e6b98d34be49114638 : SUCCESS in 9m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/02138ef365134836a7c200166ae73a39 : SUCCESS in 1h 17m 36s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/832a395bf4c147538372f68f9896821e : SUCCESS in 20m 37s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5f6614c6dcc943b68f0a24ce048fbca9 : SUCCESS in 20m 22s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/5a6a8edf668148e59cb92912d4cb01e9 : SUCCESS in 43m 45s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/4b08c785b35d473c807bde80c85181db : SUCCESS in 34m 34s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5866293f8f984e93b015d82344f71efb : SUCCESS in 30m 16s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c3dce9de15fb45fca9aab8f28c1ffc15 : FAILURE in 38m 24s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cd862510d6464eb28b54ad78f42dfc09 : SUCCESS in 53m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/aa94a4e4765e4929931a6042381c7617 : SUCCESS in 36m 22s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/19b55d8496c2463886cfaec289d9fb6a : SUCCESS in 45m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/609a385b20e24e42ad61aa701cb52c37 : SUCCESS in 27m 48s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/2f8ebc97e342400d9bc244276d682d6c : SUCCESS in 30m 18s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0ae9a56e8ef84bb7bcbb95bc810d4d40 : SUCCESS in 5m 12s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/12f0ae6117a449beb0be60039d7545d5 : SUCCESS in 46m 11s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/543f0ad27661445291ee128ba8341b07 : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c20adb2522f36de9c66fef351be99ed8d1f1f706","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 09:29:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/ed97dd5f593e45d5979979b31a92573b : FAILURE in 43m 15s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/2e86563a81324e98821b9c955eede67e : FAILURE in 43m 20s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/2012325fcdba45feaf3bbc509efce094 : FAILURE in 42m 51s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/a9a4d8468bae49c5b471d42cdaf81c91 : FAILURE in 43m 04s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/f033c8a73e1d48eaa6cc91456b8d117c : FAILURE in 43m 51s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/9550f8f014ee41669c2c1399f727dd11 : FAILURE in 44m 05s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/f457c3084f2f4894b043b54ac7580a93 : FAILURE in 44m 18s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/d3617be8177945b990e61511a0a09c2a : FAILURE in 39m 08s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/93aa4e5d591744bdbcb9d53b76316365 : FAILURE in 39m 41s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/7c3f6d50e89347b28d67bc8bb0134a95 : FAILURE in 39m 23s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/5c3c1c82d303463694bec0d465df18fc : RETRY_LIMIT in 3h 01m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c269c382356dd55f241540b4ced24b3c84cda79","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 15:48:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/c7eeef73bbb74da5a3d42b0fde316ab1 : FAILURE in 16m 30s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/f49cac34babe48efa57711de23466a07 : FAILURE in 17m 12s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/79da6ca52aa84795aa1d219acf5f8dd3 : FAILURE in 17m 12s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/d8c44d348fbe477eb6c53493747c7931 : FAILURE in 17m 03s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/e6caa79722b04d67849c2d049aa3f5cf : FAILURE in 17m 01s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/60fc2269808545718f75e3ee6a85c8f7 : FAILURE in 16m 33s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/57e003b144ed4bc491349b1391615f40 : FAILURE in 17m 26s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/95bfd99fcc7c4c6eb4b478f0017d4496 : FAILURE in 16m 27s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/d3f63ccde4a84b83a3ed49b4d96c6171 : FAILURE in 17m 05s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/c29184fd29e94dadba126c337dc61c4c : FAILURE in 17m 22s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/4065f69249454a38bedf7130639f8e25 : FAILURE in 15m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"10e234cc4458ff21a95c4009332f11a593edfe82","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 17:51:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/dc90b21b46484b7c852ca9db68fe5222 : FAILURE in 43m 00s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/807f8883341d4133b01a016c99fd4ae9 : FAILURE in 44m 19s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/39c5af187c5c4e61b20a3e8874a30fd5 : FAILURE in 44m 20s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/17cc2213b78f4be9a3eca54277042ad9 : FAILURE in 43m 53s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/8a7a1b255b044af381ceb8e2e4fd55f0 : FAILURE in 44m 25s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/8179778d667e494fb14a7893f5dacea6 : FAILURE in 44m 17s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/5f50cf3145e44114a69f4eeaf159cd6f : FAILURE in 43m 28s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/bc68450fc0bc4afbad2a63fede7c181a : FAILURE in 38m 48s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/fd0c34d77aeb46daa889d28092818b66 : FAILURE in 38m 45s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/d3528136d2cf4cd9895bd65093f9ec6a : FAILURE in 37m 56s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/3a3e1d487aba46c8a23050a2fb15bcd3 : FAILURE in 37m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"e82bb31d66a65fa2b26075ef265e4e13454b434b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 21:08:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/414c84c4f4ae4f2ab869c376935a6b5e : FAILURE in 43m 11s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/5c7651fff4044449b9fef60ee54a6fef : FAILURE in 44m 24s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/083ebac37a2740c7ac276be6a1958908 : FAILURE in 42m 44s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/d201da19c61047fe9504596b3c9e0ae8 : FAILURE in 44m 00s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/fbdad49333594a2aaed6cbe6ed8106c3 : FAILURE in 44m 08s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/188d2fd7a5c047ea8fb5464980f7e666 : FAILURE in 43m 34s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/0615e643f8ea428c94c44af03cbda0be : FAILURE in 43m 40s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/7aeba14ce2d54b02b38abaaecafaed92 : FAILURE in 39m 37s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/0f58d71b59624f5782c03bc840ccd110 : FAILURE in 38m 15s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/36602240c3da47ad927299c61dcffe18 : FAILURE in 38m 24s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/04f4c7e75e86413e8e738a856ed89972 : FAILURE in 38m 11s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"e33020f5f6aeda8d00f68f7433ab4ad7ca29259d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 23:06:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/ac7a96e237c54ef18d45e972035fbee3 : FAILURE in 41m 49s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/11c0c12abfc84cedab0a592e5b23a7e5 : FAILURE in 17m 51s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/5e189b319661437699083314cb02d3ef : FAILURE in 16m 54s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/49700045ca1d424a9283329c5db0f571 : FAILURE in 42m 13s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/98d03955870241f78533f97ec53d8377 : FAILURE in 17m 07s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/c70e74f53ea140f29b450cfcd7b36ee3 : FAILURE in 17m 46s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/8708b18a302d4bc0a0aaeed437f5263d : FAILURE in 17m 31s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/ad800812f34645c08a82679664696bbc : FAILURE in 16m 47s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/04028d6e7761436392d1a0132ac0b836 : FAILURE in 38m 35s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/4a818245e89947d9a8484b660c2dc806 : FAILURE in 15m 45s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/371138eeca1647cc9746af830ef98b50 : FAILURE in 38m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"d25d0a37a6a7402e48d6b62d4cfa9139e8216435","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-07 18:02:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/7c07c92fe1134a60aa38f0b3a7677c22 : FAILURE in 42m 22s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/0ec1e9ecdd3d4197b8cfbf317dd4bd41 : FAILURE in 42m 44s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/76cf7285c5364c7aaebb1d3605864926 : FAILURE in 43m 34s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/54f3920cc5ca484a890382a4b8aedc07 : FAILURE in 43m 03s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/da381246a0a14f20a714dcdeafcc63b0 : FAILURE in 43m 02s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/1c09fe5794a0453186988a887ebc897d : FAILURE in 44m 01s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/c7c0b64839b847939cabead42deff154 : FAILURE in 42m 43s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/cda800fc2c364e21b822a2e814fbf08f : FAILURE in 39m 37s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/dae9c86301924cb8895176e237f43a35 : FAILURE in 38m 46s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/31477035e2cf4cfd96d91b10bda1560f : FAILURE in 38m 57s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/516376cca6924efb8ca68aeb73d917e1 : FAILURE in 38m 21s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2544721ecc1cdb282acfb8e398bca069c83230a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-07 20:41:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/4c3a935090d642e88965f4f7ae4ff617 : FAILURE in 42m 59s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/0ab466758f1544d8b5005b8a57bc45f9 : FAILURE in 43m 24s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/e482a5492c6d46aeb0aefb6a104f702f : FAILURE in 43m 37s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/b757919649e04159958c4d681b72cd52 : FAILURE in 43m 41s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/a8c55956081245538beeb7674dba08ec : FAILURE in 43m 53s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/c80463fd04404a33bce5f7d2fb857edc : FAILURE in 44m 05s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/1d94a31bcab0440db3b1b920f57f6f8c : FAILURE in 44m 19s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/2f03f912092f42ef9c40465c308c5181 : FAILURE in 40m 33s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/10eb2cf529f74f84801972c20ba75c0d : FAILURE in 40m 05s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/93acf9ad3eba465189f0621cf4b8a155 : FAILURE in 39m 23s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/11124980982e4cc896d351f1a61a29e5 : FAILURE in 38m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"e563be94241c1b969f47cb4d8a3d8e0deb038407","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"date":"2022-04-13 20:13:08.000000000","message":"Abandoned\n\nCommanded not to fix","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3753e5e3cffdd19cac15ebb5d784c30ec755edb","revisions":{"a3753e5e3cffdd19cac15ebb5d784c30ec755edb":{"kind":"REWORK","_number":1,"created":"2022-04-05 22:00:30.000000000","uploader":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"ref":"refs/changes/99/836699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/99/836699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/99/836699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/99/836699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/99/836699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/99/836699/1"}}},"commit":{"parents":[{"commit":"36f31050169cc33c23a3edcae9335952065b64e0","subject":"Remove redfish cache entry upon errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/36f31050169cc33c23a3edcae9335952065b64e0"}]}],"author":{"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","date":"2022-04-05 21:59:30.000000000","tz":-300},"committer":{"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","date":"2022-04-05 21:59:30.000000000","tz":-300},"subject":"[DNM] Testing Dell CI","message":"[DNM] Testing Dell CI\n\nChange-Id: I996e5d1507363371860fe8d60dd047bc362dc8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a3753e5e3cffdd19cac15ebb5d784c30ec755edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a3753e5e3cffdd19cac15ebb5d784c30ec755edb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
