)]}'
{"id":"openstack%2Fironic~836770","triplet_id":"openstack%2Fironic~bugfix%2F20.0~Icaffb5d346001d62e5c9522708dd44f01faeb3a6","project":"openstack/ironic","branch":"bugfix/20.0","topic":"allinone-bugfix/20.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-04-06 14:33:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icaffb5d346001d62e5c9522708dd44f01faeb3a6","subject":"Start API after conductor is started in the combined executable","status":"MERGED","created":"2022-04-06 09:12:48.000000000","updated":"2022-04-06 14:59:50.000000000","submitted":"2022-04-06 14:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836770-allinone-bugfix/20.0","cherry_pick_of_change":835098,"cherry_pick_of_patch_set":1,"meta_rev_id":"8e5dec36572984f35c5fe32e465d99f316c1138e","_number":836770,"virtual_id_number":836770,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 14:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-04-06 12:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-04-06 12:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 09:14:50.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-04-06 09:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 12:11:53.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-04-06 14:59:50.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"18aee9a5ff8483cf4cce2554eb85e99d96d2985c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 09:12:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8ca56678d0ee12af51f435d9b06a1ee9442be244","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-04-06 09:14:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5ba9d0430960a5caf72aed7d6dd72264076b5f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:30:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de6e9c1636634a4182abc680d7bf45ce : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acaf72f5274a4e07850d6ed40e3f93dc : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec160a938f344e8198ad69069ac0524d : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/956fbd1eaf19476995871ed1143d6281 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6980fb2a9dd47348b5c52f50cdbfe04 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3800443959e4462f884a9bc306ff572a : SUCCESS in 6m 55s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/9631122af474436c8b49ca4a65173226 : SUCCESS in 12m 09s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/aeccb09653d647f991b14f05c1a08410 : SUCCESS in 9m 50s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/3d3bc0c72d384a408e8ff1cd6d57419c : FAILURE in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ae98b7086d7055dcae170b14adcb171157e307e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:52:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/208169d1f28b43a09c17ca57ce2257c6 : SUCCESS in 15m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/926ebd9168f04562be13f440755e8147 : FAILURE in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2630c506f1db114cbe36ae6898fa63b6098362bf","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-06 12:11:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcba914a6b8544740e7c9e0214715a701de3eb12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:12:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f18e754e1c556d28c6be19333f72772747fcacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 14:33:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1312ed3a599a4a69bc951244ecef47d0 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93f585a1b15541f39f0316c436b0beac : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84a8a05ee31341728e7b1d8e311129ea : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118356a635e94f6bba6fe41c2d4cff0e : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13528492bf24a259e20e1bec2a1bc75 : SUCCESS in 8m 09s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2e30a0da175943129d88f185197c25d4 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8eb820f68d9f623bf7402f3bfeacf902efbc8473","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 14:33:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be220ca208a03fd89e06e047c114fe773681a41f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 14:35:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a024ccdb314642ab23def1d053d854 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66d17eda01b14c5d997bba93089e51fb : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8e5dec36572984f35c5fe32e465d99f316c1138e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-06 14:59:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- R650-bios-idrac https://stash.dellemc-community.org/t/openstack/build/5c5cb5705b504c8aabbe174dcd21e4d8 : SUCCESS in 57m 44s\n- R750-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/39ac862841884c4889826f59237f1e25 : SUCCESS in 51m 29s\n- XR11-bios-redfish https://stash.dellemc-community.org/t/openstack/build/e1851fbe37ad47fe811fc504e9a33f57 : SUCCESS in 52m 33s\n- XR12-uefi-idrac https://stash.dellemc-community.org/t/openstack/build/7b7c4b729bf64840b428c7a4177fc864 : SUCCESS in 54m 29s\n- 14G-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/909da87528264568bfdef6fac8c76aac : SUCCESS in 51m 01s\n- 14G-uefi-ipmi https://stash.dellemc-community.org/t/openstack/build/636cca18d050439bab3ceaacca6c8695 : SUCCESS in 52m 00s\n- XE2420-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/bb28447c8fa8445a846200a3e69d5735 : SUCCESS in 50m 32s\n- R7515-uefi-redfish https://stash.dellemc-community.org/t/openstack/build/a39f48a83b1542d7973b934e324ea8b6 : SUCCESS in 52m 14s\n- 14G-bios-idrac https://stash.dellemc-community.org/t/openstack/build/44dcb48ba447403f81e61246c6491942 : SUCCESS in 52m 49s\n- 14G-bios-redfish https://stash.dellemc-community.org/t/openstack/build/74f8c1b4ff954caf88133f3b2af24ae1 : SUCCESS in 47m 21s\n- R6515-bios-idrac https://stash.dellemc-community.org/t/openstack/build/b5c6f0279e534b869b607ae3275dd61b : SUCCESS in 2h 57m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7708e3d9ff993e222889048ff0d94450056cc3fe","revisions":{"7708e3d9ff993e222889048ff0d94450056cc3fe":{"kind":"REWORK","_number":1,"created":"2022-04-06 09:12:48.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/70/836770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/70/836770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/70/836770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/70/836770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/70/836770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/70/836770/1"}}},"commit":{"parents":[{"commit":"750e31abd8d9adcab0ab033f650335f4910690a2","subject":"Fix rebuilds using anaconda deploy interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/750e31abd8d9adcab0ab033f650335f4910690a2"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-03-24 17:07:51.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 09:12:48.000000000","tz":0},"subject":"Start API after conductor is started in the combined executable","message":"Start API after conductor is started in the combined executable\n\nChange-Id: Icaffb5d346001d62e5c9522708dd44f01faeb3a6\n(cherry picked from commit c5ef4b9203d396d02dde92760acd95a0912caa68)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7708e3d9ff993e222889048ff0d94450056cc3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7708e3d9ff993e222889048ff0d94450056cc3fe"}]},"branch":"refs/heads/bugfix/20.0"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
