)]}'
{"id":"openstack%2Fironic~839739","triplet_id":"openstack%2Fironic~stable%2Fyoga~I6a8f38934b475f792433be6f0882540b82ca26c1","project":"openstack/ironic","branch":"stable/yoga","topic":"story/2010016-stable/yoga","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-05-10 18:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a8f38934b475f792433be6f0882540b82ca26c1","subject":"Exclude current conductor from offline_conductors","status":"MERGED","created":"2022-04-28 16:57:29.000000000","updated":"2022-05-10 18:48:56.000000000","submitted":"2022-05-10 18:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839739-story/2010016-stable/yoga","cherry_pick_of_change":839704,"cherry_pick_of_patch_set":2,"meta_rev_id":"c89f645cb8d12aef6ea6278ca2a8759a236b2c6a","_number":839739,"virtual_id_number":839739,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 18:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-05-04 13:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-10 16:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-10 16:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 16:58:50.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-04-28 17:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-28 18:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 18:57:47.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-05-04 13:38:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-05-10 16:50:40.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"1dafea5a111d0852a32fc899618e2fc4b9e02c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-04-28 16:57:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8ae39a0d80238b5ae135d30c5bcbf252e4afc21","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-04-28 16:58:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"25b59e53188dae6cb1838a200babb5a376882f1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:20:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a522225f884644438f3774d37df68532 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6d0586e7c2e43e99bced5e10ecf7453 : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c1a2601b1f10b69d956258dde0bd7b9d55d6f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 18:43:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a19612cd48b4afd9fefd128893f84e2 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa6e3564d14451bb4a3b6868e816585 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99ab2f490f424e9c89abad1478fa2991 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e0a66c8c4646fcb09e5f37661ad96a : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0bc67a7a261470196c4473f5a181d3f : SUCCESS in 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e06f6da445c1411c99a45a79c77214a5 : SUCCESS in 9m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/bce2c92927f143938ee4bb7ca7fd32c7 : SUCCESS in 11m 50s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3fad4f89e29644bb9ba50117918a6725 : SUCCESS in 10m 05s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cc4029dfd02747559c884215a698f63a : SUCCESS in 26m 24s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/8bd0c26f768b4c60bbf8f313ef992c36 : SUCCESS in 25m 37s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/50af9d78710b4913a3b7eaa739034227 : SUCCESS in 1h 02m 21s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/4b937983d6b54c37b1830fe14562b2a9 : SUCCESS in 1h 14m 27s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/dd355906b631438884ff1d52840a0064 : SUCCESS in 51m 05s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6b8f872643c04c749c18d679baa42865 : SUCCESS in 48m 35s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/49d66762b04f48e5aedbc71e7c3dda43 : SUCCESS in 52m 18s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ac02589b352140248489fab5a04ed3cf : SUCCESS in 41m 06s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/95deafae8ef04441a1a5982567758990 : SUCCESS in 1h 26m 04s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/af9087db80ed49ebaf1fc4089d33b690 : SUCCESS in 50m 50s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/fd3c113a8c924ad498fa4dea0231d0ac : SUCCESS in 1h 43m 43s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9b5e76c9f3a04e2d890027855be2c8af : SUCCESS in 55m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7edde64340dc4e9a93423f0f165bb931 : SUCCESS in 36m 52s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/1cf0fe8acdba4eb0a482f04de4491062 : SUCCESS in 46m 42s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/c119d479ec2c4e31a46a336e868db6c3 : FAILURE in 27m 08s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ab7884ca7cda47d3a62fc217150e72ac : SUCCESS in 5m 13s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/fa1cd8e0097a4df699d68ebf64112c2d : SUCCESS in 55m 14s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/2f470eabee4540f89c6d1a9cd5ccc58b : FAILURE in 1h 36m 05s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b1c8c5656d654f5087b6ec8590ffe05e : FAILURE in 1h 00m 51s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/09199153f3964a99b2e16732461d02e3 : SUCCESS in 43m 16s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/18a93c12f77f4cad92e0ccd973277f71 : SUCCESS in 56m 55s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54d2ab628f2b4c888699887c321d4b92 : SUCCESS in 31m 29s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/9c301f0526324fab898f6fd91c90790e : SUCCESS in 28m 45s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/e4ff3875a8214430aedef978edc7cb70 : FAILURE in 1h 23m 45s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/ba80c582b9f14876aae3e39e232cea44 : SUCCESS in 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d035582075e6b78ad1b7f1985cb168096fe68beb","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-28 18:57:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- R650-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/be6cebe3cdc244e58da09ead21ba0c63 : SUCCESS in 59m 45s\n- R750-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/0444de8f8aef46db8f2de47dbf629be8 : SUCCESS in 57m 23s\n- XR11-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/b639e20ec86347729125b693c30f104f : SUCCESS in 58m 49s\n- XR12-uefi-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/9ecee33cdcaa4cc99888a6b5f5d323bb : SUCCESS in 1h 00m 12s\n- 14G-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/abc9ea75608b472686a0281206969e2e : SUCCESS in 58m 51s\n- 14G-uefi-ipmi https://stash-dellemc-community-org.us.dell.com/t/openstack/build/2ce0991aff0e4648a4a3dd48efcb1a32 : SUCCESS in 58m 48s\n- XE2420-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/4de06521797e4bb99839de055e55a430 : SUCCESS in 58m 14s\n- R7515-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/82baae50e7694624b6e98fc314154326 : SUCCESS in 50m 48s\n- 14G-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/a24b8e09007b4cf49937e4a59aa0fbe5 : SUCCESS in 51m 45s\n- 14G-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/a1f85436c3da4babb40914b28b3e2b53 : SUCCESS in 52m 52s\n- R6515-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/39f2cc453cd541b697248840c3ee2670 : SUCCESS in 58m 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"8e5f21a46bc17bba5038016f510a96691171153e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-04-29 13:50:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0ccc24076b6eec282fc947898cf67ef93a034dc6","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-04-29 13:51:52.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"9016788e05c7ed13f28788cce1dabae6ac85ee54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:10:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b72ceed5350418aabf2a820b6d92519 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaa3795a704c4fb5aff5c6d8212edbf4 : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"996f44b7b87a55dfa17f53c242f5de9725f636d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:45:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/539818904f9e4b7bb3777472e9c9bf7b : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022fc541e03b49b281bc3e9c7a0c60cd : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6489389627644e1f822ff3a227f3789b : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98027255d95f44449326bf8b26a686d5 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b12cf8fd94a4deb98f6b6a6f7301363 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79f3a0d1d12940c2a26f857e16ff2bbd : SUCCESS in 8m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/70509123c96442dcabfc60e255f5eece : SUCCESS in 12m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b597ec108b114617953213bf71eb44f5 : SUCCESS in 9m 52s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1a7209b2a428477f8f6ad9ae039b7b4d : SUCCESS in 28m 13s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0c0ca1d4721442d59edbb0a66d7063ec : SUCCESS in 23m 05s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ada8b4c90c934460b86246f7f3206016 : FAILURE in 47m 29s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3511cfcb780c4da691a8bd7b0f5551ed : SUCCESS in 1h 03m 16s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/efbd9ad5da774082861de0c421111acc : SUCCESS in 1h 02m 35s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3af2355df03f433e842be420893156e1 : SUCCESS in 1h 06m 39s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/237c0ccd78fb408b9787e4655a291a96 : SUCCESS in 56m 21s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e1e935d7a95944e8b7617bc8921616e0 : SUCCESS in 39m 23s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8fdd682419bc49c7940805e92d197981 : SUCCESS in 1h 22m 21s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2c6aadc06811495082ff1f54f27d861e : SUCCESS in 53m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7cd65ddaa27e457fb12163843a5973f9 : SUCCESS in 1h 51m 51s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e6ae06e74d464c4c8a1a6d0f21a3ee59 : SUCCESS in 1h 00m 10s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/08cc455d63144135acaa7107a987379a : SUCCESS in 41m 25s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/dce967db966842a8aa3faa79b2715e97 : SUCCESS in 48m 50s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ff658f58051944ae983b494cfe9b9399 : SUCCESS in 1h 27m 28s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/e6463985c5eb445eaf3d9811f464c021 : SUCCESS in 5m 10s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3557b48afb63495b8aebab15610d6fa7 : SUCCESS in 53m 37s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/b0ae99dea32145c88c831fa59ba26675 : FAILURE in 1h 29m 51s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8b701b22b01145fbaa1dd6b989ed4f20 : SUCCESS in 1h 02m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d9d7f5a63d344925b1d30958131cbfe4 : SUCCESS in 43m 06s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e6fd35e6f78a4c3cbb10e51a0de300b0 : SUCCESS in 58m 12s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74bcb51298334974975baad93c51c3e0 : SUCCESS in 25m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/8449b619c725440d9b2be4d1c13b0cd4 : SUCCESS in 39m 42s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/5fc97ebc1f76480d868db536603afeba : FAILURE in 1h 22m 39s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/fa4d8490bd19432b876fcfe4695ab38f : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f0e0a3cd2db38362b8bd6dcc61a68e038dc50c2","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-04-29 15:49:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n\n- R650-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/8fbd4747d8384443b6d0f55214aceaed : SUCCESS in 1h 00m 58s\n- R750-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/d8ed7c92e9504509b0534408d866dd7a : SUCCESS in 56m 05s\n- XR11-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/1546725d9c82433abe50b3875f5d4c77 : SUCCESS in 1h 00m 06s\n- XR12-uefi-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/cf21e6ee6e864cf5beeed0503b3c8c35 : SUCCESS in 1h 01m 26s\n- 14G-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/c50ebb7725194dbda5866db3fa1547bd : SUCCESS in 59m 46s\n- 14G-uefi-ipmi https://stash-dellemc-community-org.us.dell.com/t/openstack/build/2b410a48a8fe4e24b67ab423c4400a2f : SUCCESS in 58m 49s\n- XE2420-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/8981ea0bb630475680f286459911aac8 : SUCCESS in 58m 56s\n- R7515-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/90f7069da5f84699a4b09874612c5e3d : SUCCESS in 49m 59s\n- 14G-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/cb3f1b608cfd4fdcb2dcfa65c3e2b89f : SUCCESS in 50m 22s\n- 14G-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/a10a51b9af284e6e9d38ef34379afd0a : SUCCESS in 52m 44s\n- R6515-bios-idrac https://stash-dellemc-community-org.us.dell.com/t/openstack/build/0262d2eb66384c1f854d6353e5abca14 : SUCCESS in 56m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"d046a2ad7a8bfd7ba86a8763e3fa40999e8d2e6a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-04-29 16:37:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"988510b52e7be9ec3d5b24eb0a9ebe0febe38eae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:22:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef821545788243e685d25808442a7de3 : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e917119fba4b4b7f9410d9ef607a43c1 : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d2944458959f7c5e10a6b46302e2e04cc93d77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 18:23:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeafa026a0f2408599958885b1db3a69 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbdb04a67c743e581573358e69c885f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8baaa224a98941c1b81c230a44131c06 : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3784fdfdd10446880c2e5ce14960fbf : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d2189d38254de98d4793fd1ba5471c : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc513b43e4944aa8ab9cd5fd970ea07 : SUCCESS in 7m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e18d6dff352d4282b052d8d60746f717 : SUCCESS in 12m 43s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/000692a44c514fcbbd7214661a87ea3f : SUCCESS in 9m 35s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/7375b6f85faa40888776c7b95f4344a4 : SUCCESS in 32m 13s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/91aad5d61b1940ae9046d8c7fc8b0a98 : SUCCESS in 35m 41s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6469184be61041ac9e9b48c9358b72a0 : SUCCESS in 1h 20m 59s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d2014fe2f7cc4b4187c00ba627e4310e : SUCCESS in 1h 01m 01s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d895be1fde79461890ee2a52e2069b2e : SUCCESS in 1h 07m 46s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9ff130358bae4c5586cc5b15628aefb4 : SUCCESS in 1h 06m 19s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/31819da0a9a44a29a13ed39817b8b8bf : SUCCESS in 53m 39s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/26d698bd225b43e5b20a1129c0a6684b : SUCCESS in 48m 34s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/bb481a8e9f94438bad4017dbdb130f6f : SUCCESS in 1h 25m 20s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/e018feea0cfe496e8815b4d7911696bf : SUCCESS in 59m 19s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/38b15d13cfb64db4ad693ae93519b561 : SUCCESS in 1h 38m 34s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/2b7c504230494534adcf7ddad18fcf40 : SUCCESS in 1h 03m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ead70ae8add249a7a6f7b7413d635c94 : SUCCESS in 42m 53s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a5f3f255d16c46648b75e67a332d14bb : SUCCESS in 54m 17s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/664fbefbd2164f09a25b47be20637445 : SUCCESS in 1h 39m 42s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/aac0a93b43584f99945386c3e120fe42 : SUCCESS in 4m 09s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e975cfedebed4f48b90528e365f0fccb : SUCCESS in 56m 12s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/6f25ad8bda38481da53bd89671938c85 : SUCCESS in 57m 52s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2fde12de7728459c87325770aa176606 : FAILURE in 55m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/08315bc107f54426ab4fce246a00a93e : SUCCESS in 41m 36s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e9e8caf4161a46f38573c5788ad90e0a : SUCCESS in 49m 49s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afb15893065a4cfe9bbe898e98fa6778 : SUCCESS in 25m 12s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/3d5baef95fd14de88c2a9ba9b5e3e381 : SUCCESS in 29m 19s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/898b9ee2e954406e9a9bc63fd1b23c48 : FAILURE in 1h 19m 39s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/d26ba5ff26b14397b3f27cf18826d703 : FAILURE in 47m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f64a55aefab9440bf51f0e9dff5488167e8856c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-05-04 13:38:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3dc0e9d336c8631e71a58aa51bee7af2291f320e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-10 16:50:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67ed75138867e2d8be71e4b5a243434ea920b82e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 16:51:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58bc68f6b4367273a4f2ab28d2f174aeb11df60b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:47:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdbbba64a129463eb4c5b85097206e4f : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b41964e63344dc4a0384959d17d7c07 : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa80a990cb1344eb84491f9986a26a56 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d97c1d0a72a48bb99ae5131a38deb50 : SUCCESS in 14m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e27ac6009ba435e9faa8b033fe631bd : SUCCESS in 7m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ffff9e44d0d945fdbf592966a0302b20 : SUCCESS in 12m 31s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e72f7863da4b4f5eab3e4cd0e5ebe2b6 : SUCCESS in 32m 24s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a197f95cba544dea96c283f8b1d6c188 : SUCCESS in 33m 34s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/44eee9f25a30434c8e8ca1b89dd1330a : SUCCESS in 1h 00m 23s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/fe5fda26cab942e590d2c2d2fa4c2fab : SUCCESS in 1h 11m 39s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0180924f76414a6f96dfe1be2ad670c2 : SUCCESS in 1h 04m 45s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/be7d036439f04d4f8be1fa30eef612d1 : SUCCESS in 59m 32s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e695c202326945be9b5d8c7d042091a3 : SUCCESS in 57m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/756dff41ae8a477cb4d5a65858b4729a : SUCCESS in 50m 46s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/758eaefa789b4f978c3703cdef2b6614 : SUCCESS in 1h 24m 10s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/c7395fa352444521a96f26ac78af0fbb : SUCCESS in 1h 02m 39s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/90ac684047fb4b52bcc1cb2df44e0e91 : SUCCESS in 1h 09m 59s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c38a8c94d05a4f7fb4178369960ef518 : SUCCESS in 44m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e870be2ecc61418ea983964d6f762002 : SUCCESS in 1h 01m 38s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b3b8dcf223ca4e8a99a4fff5433cdf89 : SUCCESS in 1h 50m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ae7596eda2e933f2b5ecadc2125cfd930d1eae61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c89f645cb8d12aef6ea6278ca2a8759a236b2c6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:48:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea26b5c8ae94b4e8d616f25bce83c49 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f76d2a7372734613881f280be8cc3acc : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86e326a59126a24024649e8454b9c9253206c58b","revisions":{"fbd62da750263b3b3288579839a9e27eba65ec89":{"kind":"REWORK","_number":1,"created":"2022-04-28 16:57:29.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/39/839739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/839739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/839739/1"}}},"commit":{"parents":[{"commit":"e5626ab9c7d1b71f078b90e4c785a4aaff6d05b3","subject":"Start API after conductor is started in the combined executable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e5626ab9c7d1b71f078b90e4c785a4aaff6d05b3"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-04-28 07:22:14.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-04-28 16:57:29.000000000","tz":0},"subject":"Exclude current conductor from offline_conductors","message":"Exclude current conductor from offline_conductors\n\nIn some cases the current conductor may have failed to updated\nthe heartbeat timestamp due to failure of resource starvation.\nWhen this occurs the dbapi get_offline_conductors method will\ninclude the current conductor in its return value.\n\nIn this scenario the conductor may end up forcefully remove\nnode reservations or allocations from itself, triggering takeover\nwhich fail on-going operations.\n\nThis change adds a wrapper to exclude the current conductor.\nThe wrapper will log a warning to raise the issue.\n\nRelated-Bug: #1970484\nStroy: 2010016\nTask: 45204\nChange-Id: I6a8f38934b475f792433be6f0882540b82ca26c1\n(cherry picked from commit 4cf0147e86039e8fa2232c3de6d45f1405a515d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fbd62da750263b3b3288579839a9e27eba65ec89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fbd62da750263b3b3288579839a9e27eba65ec89"}]},"branch":"refs/heads/stable/yoga"},"86e326a59126a24024649e8454b9c9253206c58b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-29 13:50:26.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/39/839739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/839739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/839739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/839739/2"}}},"commit":{"parents":[{"commit":"e5626ab9c7d1b71f078b90e4c785a4aaff6d05b3","subject":"Start API after conductor is started in the combined executable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e5626ab9c7d1b71f078b90e4c785a4aaff6d05b3"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-04-28 07:22:14.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-04-29 13:50:26.000000000","tz":0},"subject":"Exclude current conductor from offline_conductors","message":"Exclude current conductor from offline_conductors\n\nIn some cases the current conductor may have failed to updated\nthe heartbeat timestamp due to failure of resource starvation.\nWhen this occurs the dbapi get_offline_conductors method will\ninclude the current conductor in its return value.\n\nIn this scenario the conductor may end up forcefully remove\nnode reservations or allocations from itself, triggering takeover\nwhich fail on-going operations.\n\nThis change adds a wrapper to exclude the current conductor.\nThe wrapper will log a warning to raise the issue.\n\nRelated-Bug: #1970484\nStroy: 2010016\nTask: 45204\nDepends-On: https://review.opendev.org/839910\nChange-Id: I6a8f38934b475f792433be6f0882540b82ca26c1\n(cherry picked from commit 4cf0147e86039e8fa2232c3de6d45f1405a515d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/86e326a59126a24024649e8454b9c9253206c58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/86e326a59126a24024649e8454b9c9253206c58b"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
