)]}'
{"id":"openstack%2Fironic~843523","triplet_id":"openstack%2Fironic~bugfix%2F16.1~Idf7991d8f8a8cebcdc0413f26ad31542f8ffa8a8","project":"openstack/ironic","branch":"bugfix/16.1","topic":"drop-queue-ironic-bugfix/20.2-bugfix/20.0-bugfix/19.0-bugfix/18.1-bugfix/18.0-bugfix/16.2-bugfix/16.1","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-07-13 14:07:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf7991d8f8a8cebcdc0413f26ad31542f8ffa8a8","subject":"CI: Removing ironic job queue","status":"ABANDONED","created":"2022-05-26 20:25:39.000000000","updated":"2022-07-13 14:07:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":843522,"cherry_pick_of_patch_set":1,"meta_rev_id":"0c2546f30bc547d41f8578d90d409322b0376b9f","_number":843523,"virtual_id_number":843523,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-26 20:50:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-26 20:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dd46cc5e17d56a851d9e23ccbe8ec115b9ad34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-26 20:25:39.000000000","message":"Patch Set 1: Cherry Picked from branch bugfix/16.2.","accounts_in_message":[],"_revision_number":1},{"id":"e3417bf9c4a5986196d866ee4fc13ce497afefd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 20:50:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb38a21b3d1445b5a1a98d46003bb422 : FAILURE in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fed5439a0c4b67b6d67cfc32985b7b : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87c097d2031b492cb4a6171df68bf41d : FAILURE in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4145b25e80a748a5b2faca428f7124f5 : FAILURE in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af2b73ac5564873be6d9e3ee9644e00 : FAILURE in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b957214cd1994ad8bc673236b444df5b : SUCCESS in 17m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/205892f2153f4de880bce385268c8cae : FAILURE in 9m 44s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f0b88c87b4d543feacceb2fd3d292c0d : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c2546f30bc547d41f8578d90d409322b0376b9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-07-13 14:07:38.000000000","message":"Abandoned\n\nno longer needed, branch to be shutdown.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc547b79a7b9d96a4ea75510ff4ad15726730538","revisions":{"bc547b79a7b9d96a4ea75510ff4ad15726730538":{"kind":"REWORK","_number":1,"created":"2022-05-26 20:25:39.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/23/843523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/23/843523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/23/843523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/23/843523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/23/843523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/23/843523/1"}}},"commit":{"parents":[{"commit":"418699c48b56f13a7d140ea52886b12980498e74","subject":"Update .gitreview for bugfix/16.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/418699c48b56f13a7d140ea52886b12980498e74"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-19 18:08:22.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-26 20:25:39.000000000","tz":0},"subject":"CI: Removing ironic job queue","message":"CI: Removing ironic job queue\n\nChange-Id: Idf7991d8f8a8cebcdc0413f26ad31542f8ffa8a8\nSee: https://zuul-ci.org/docs/zuul/latest/releasenotes.html#relnotes-4-1-0-deprecation-notes\n(cherry picked from commit 63e53797adca625edd2fdacd16a025bedccf8731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bc547b79a7b9d96a4ea75510ff4ad15726730538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bc547b79a7b9d96a4ea75510ff4ad15726730538"}]},"branch":"refs/heads/bugfix/16.1"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
