)]}'
{"id":"openstack%2Fironic~846639","triplet_id":"openstack%2Fironic~master~I5fc9fea1c870328953103dd6b092929da038ec61","project":"openstack/ironic","branch":"master","topic":"test-ci","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-06-23 14:51:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5fc9fea1c870328953103dd6b092929da038ec61","subject":"[DNM] Test grenade swift workaround","status":"ABANDONED","created":"2022-06-20 14:39:54.000000000","updated":"2022-06-23 14:51:23.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e12c9f2b371cbfc5bfd777f7e3102e6e9f8ce08","_number":846639,"virtual_id_number":846639,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-21 10:21:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 14:54:26.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-06-20 14:54:34.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-06-20 14:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27515a75d80f7ee25be5781deae39b1299fd7e48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-06-20 14:39:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3732ab3b3ea19b524cf30fe36f30258f12bfb6ca","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-06-20 14:54:26.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2c2af237732286ecf1d6ae86f98151072694e89","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-06-20 14:54:34.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"56d33dca40ce726d8c151d7081424d031e7d9172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 14:54:56.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d8316eedc654f1d5962ad5dc717845b726d9eb2b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-06-20 14:56:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ca8b28ee11ffbb6cb2307add3f6b4655527e167","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 14:59:33.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b237699c790de90976297b24cd49d2fc3f7dd545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d6ff9624e224ca9b09c4bd840a82b48 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b3ad66168a4f058d0103f09a9b6e60 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa5167c120e4899b0608aa82e334a31 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc9ff9220cd4293960dfb687c64baca : SUCCESS in 8m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9ab189dc56c40758343d95006e54a57 : RETRY_LIMIT in 2m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f12847fc0ed4133bb7e13811e872361 : SUCCESS in 14m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1753d61dfa06479fbc82f5e2fcb63a6c : SUCCESS in 12m 24s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1662f37a1b5c4d85b8d79fddcc43a212 : SUCCESS in 8m 02s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/81ee3497c78145d084a098e27a28f4a1 : SUCCESS in 32m 45s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/07b6c7de8ec54a4d99cc00e26b466238 : SUCCESS in 31m 53s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/bb59ecb61b3f4dfd829c6a0fc6b472b5 : SUCCESS in 1h 12m 17s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/645a0d57558b4c50bbe8c51ca68ba7f3 : SUCCESS in 56m 25s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/8246846744e34611be818932106c2334 : SUCCESS in 53m 54s\n- ironic-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/5a9532a409de414eae5e518510525af7 : SUCCESS in 56m 03s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/71b17cb1217247d1b6c05b95ac450414 : SUCCESS in 58m 20s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1150021122df44699686d3653ab3fc6e : SUCCESS in 44m 57s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/ba4fde8b184b477788aa6252a88cf5d5 : SUCCESS in 1h 05m 59s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/ca3943a711de47bca93d5dc1820d5530 : SUCCESS in 56m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/02841610452e44b6ae69775dfb34326b : SUCCESS in 1h 43m 18s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ae02a319b165487bba99567a2dd406a7 : SUCCESS in 56m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/02d521fd8ac24fb0a89abcca53e27df8 : SUCCESS in 35m 11s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/014603ad09a948e7befbfefc33f08fff : SUCCESS in 54m 25s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/fc8a7d5a61f94133b205ff9d55b11ad4 : SUCCESS in 1h 43m 54s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/863275217c844753aa0ad7568e85e238 : SUCCESS in 5m 27s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ff1cb38074a94284a01f4a1eb8afa1c9 : SUCCESS in 1h 00m 55s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/ea46696c679d4a78804f77382b81886d : SUCCESS in 57m 54s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/c928a3d8966c46dda42922fa14c19cab : SUCCESS in 43m 00s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/bb4fce1fc18543ccbab6ec36a7605615 : SUCCESS in 58m 43s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/3d5f3117dcf14be7820298f6c311ada8 : SUCCESS in 55m 08s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70c5b72145a746f7b0907389ad964f64 : SUCCESS in 28m 46s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/a94c0dbd066a412fa398261c7920c006 : FAILURE in 4m 27s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/d3450317591a4ddbadfd913ad05a215c : FAILURE in 1h 22m 04s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/b049b59f45b9455590dd59c0c5ad94dd : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92fad18e74624f9d5002eec9c132c6d256d87c85","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-06-21 07:13:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6d2baa0e475d71d56b39b910a9b35681ce8b69f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 08:19:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2b22a1b140f415ea38a1c6573c2c0bf : FAILURE in 19m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8275c13ab174c1592ed784be6efe480 : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee4d4adcf82e2803f3e822851423c69764d28262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:21:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2980643d8c2485a9b8ac6cf50bba320 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010a5cda03fa470c8787886f41551d65 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24cefebfac2b4ae9a596f13a7cf2bdc0 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b52fb79081954ae0bb87f5cdf47bee1d : SUCCESS in 10m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f27800465a0441f5b90ba2d39cba11c8 : RETRY_LIMIT in 2m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240577c7ce7a4b0fb90473fbe0160e6d : SUCCESS in 12m 35s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4eaa2f12a02a4f4191f704cc8d1497f7 : SUCCESS in 12m 52s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b510d9339d7b455abd3bc6fbaadebc9c : SUCCESS in 8m 55s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/54b9ec6e7f09498d83483c2a2c3ae10e : SUCCESS in 28m 49s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1e6fd0ca1e6b4720bb3c4ee7c0cf0f0f : SUCCESS in 36m 41s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/c8ea1f09229f4d5d9ec11d602aba6664 : SUCCESS in 1h 07m 58s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/03e80fc0f02b46d0b3f2cf4d2d0c38e9 : SUCCESS in 53m 29s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f8a0fbc5de33432493117cc55f6ca994 : SUCCESS in 1h 04m 56s\n- ironic-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/59b6cf892d464899beb24b9f42fc9b10 : SUCCESS in 51m 55s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c69c8e71d6dd446480bd77a54f5df1fa : SUCCESS in 50m 55s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/0a9eabc6f6eb43d09b813ef09a79cb7f : SUCCESS in 45m 16s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/88b6efe08c7f403e8d693878c51ed762 : TIMED_OUT in 3h 05m 27s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2f9fd3a045d84974841f6ce1232ba549 : SUCCESS in 52m 37s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/99c6e114ee6345739b2cb146cb6b81f5 : SUCCESS in 1h 41m 59s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/afcfb96026384f4dbfc83ea323aabb47 : SUCCESS in 1h 03m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/25b015b955574b7ca8408b7b873c3568 : SUCCESS in 37m 40s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7af2b02c88364ab0b21d096b575d01f9 : SUCCESS in 48m 01s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/ea6f5ef0721e49f7958a7793e0869836 : SUCCESS in 1h 47m 39s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/57c22f80e7454bc4940fb716b2d7a0eb : SUCCESS in 5m 07s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/af6d154aa0da4ce2aca2b1066c0c31ee : SUCCESS in 54m 19s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/76617d4d8ce44e04af86f34f81cbf0ac : SUCCESS in 55m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4494de13b893412e90d17ed52714e163 : SUCCESS in 45m 35s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/dce17434074c401eb3466e4fe47b2bf5 : SUCCESS in 1h 00m 30s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e1b8e1fff42746769a0f1ddfeccaefef : SUCCESS in 55m 46s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8ef802494ff431cb29063d909d145be : SUCCESS in 26m 08s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/33c1821559a24432a29e3159e09bb914 : FAILURE in 3m 53s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/eab7f0a3044a42528bd588d5522a2c58 : FAILURE in 1h 20m 20s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/72216efc60df456c8f1a00d6594c1851 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e12c9f2b371cbfc5bfd777f7e3102e6e9f8ce08","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-06-23 14:51:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5aaade0d90d8c0a6cc2eb9f202820faac77c56a","revisions":{"c5aaade0d90d8c0a6cc2eb9f202820faac77c56a":{"kind":"REWORK","_number":1,"created":"2022-06-20 14:39:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/846639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/846639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/846639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/846639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/846639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/846639/1"}}},"commit":{"parents":[{"commit":"39a7f58002eac8ad10720c7347f001c8afd24a6f","subject":"Docs: replace nova cli calls with openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/39a7f58002eac8ad10720c7347f001c8afd24a6f"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-06-20 14:39:00.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-06-20 14:39:00.000000000","tz":120},"subject":"[DNM] Test grenade swift workaround","message":"[DNM] Test grenade swift workaround\n\nThis is a test\n\nDepends-On: Ic19030028376df9e3b0ddf2fa81932686707e9b1\nChange-Id: I5fc9fea1c870328953103dd6b092929da038ec61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5aaade0d90d8c0a6cc2eb9f202820faac77c56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5aaade0d90d8c0a6cc2eb9f202820faac77c56a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
