)]}'
{"id":"openstack%2Fironic~851413","triplet_id":"openstack%2Fironic~bugfix%2F18.1~Ia2ce34f09695731b0f48798f662006c4904e2223","project":"openstack/ironic","branch":"bugfix/18.1","topic":"redfish-raid","attention_set":{},"removed_from_attention_set":{"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2022-08-04 09:48:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_27909\u003e by clicking the attention icon","reason_account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-08-04 08:34:56.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":[],"change_id":"Ia2ce34f09695731b0f48798f662006c4904e2223","subject":"Fix Redfish RAID for non-immediate controllers","status":"MERGED","created":"2022-07-28 14:18:26.000000000","updated":"2022-08-04 10:51:52.000000000","submitted":"2022-08-04 10:50:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851413-redfish-raid","meta_rev_id":"8bb9da3470ed924813bf6a2b2af5eee5f53ee787","_number":851413,"virtual_id_number":851413,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 10:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-08-03 21:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-08-04 08:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2022-08-04 08:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 14:18:26.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"CC"},{"updated":"2022-07-28 14:33:21.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-07-28 14:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-28 15:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 20:43:22.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-07-31 21:35:05.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-03 21:23:25.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"f863f652fe2b4cb838528970066d637aebe6e865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-07-28 14:18:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f7bf874b8301b631763b8eb1c49c2433d67414a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-07-28 14:18:46.000000000","message":"Topic set to redfish-raid","accounts_in_message":[],"_revision_number":1},{"id":"c74016680d2164c2ef43637e03e0fe6d479bb06a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-07-28 14:33:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/03b7802/03b7802.html : FAILURE in 5m 55s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/a16c338/a16c338.html : FAILURE in 6m 06s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/cfd7e05/cfd7e05.html : FAILURE in 6m 41s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/6e9edcc/6e9edcc.html : FAILURE in 6m 35s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/b89270c/b89270c.html : FAILURE in 5m 56s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/440b835/440b835.html : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c42529219551e0c341e6b61d30b6dcb34c5baa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:43:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b280ae6f13ad4ce9a559ccd4039f0eed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58f87265406542ee9c8827af139d4980 : FAILURE in 18m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a74553b974944dd2bfbfc88304e071a1 : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"781a26035b47235ddfc478451794c957c842dc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:51:54.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ad782b4b0e8488d9c7d22e3692c41c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57856801356545aa9a4eb71c9dfdfee6 : FAILURE in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016f080dd47744b69459b42645f66ad3 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b5c3553d53140d88d3257d02d476e20 : FAILURE in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eebb44b972f244a5922272be4deef5ed : FAILURE in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81cfcff883444c60ae15c730a7562ce3 : FAILURE in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a6d9f1d2724775841565317e117aab : SUCCESS in 14m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0745cd97abef475987ada6468b8bacab : SUCCESS in 9m 04s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0252ff6f4b62431f81c58cc3309b1b1f : FAILURE in 10m 07s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5518be0fae83401fbca5c09b481ac630 : SUCCESS in 23m 54s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/6a944a6ab209434ea390ff496e378462 : SUCCESS in 1h 21m 02s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0b5f57d6b1904b709c2de780ed0cffac : SUCCESS in 48m 33s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/64cde73a21d444ee9c1329177580226f : SUCCESS in 55m 10s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2ab52b9916bc449e9ec6a72d8e27309d : SUCCESS in 59m 39s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/afde926a60e347a0b601cc4d93df3e63 : SUCCESS in 1h 08m 48s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4c0ab592ca8246a783e762b61e544cea : SUCCESS in 52m 03s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f452ab1b4f264f49b68244111c102b02 : FAILURE in 6m 30s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0cfae24604594430a7ce11977044a2c1 : SUCCESS in 1h 01m 50s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4effd5247a2243fe860eda0d01817e1b : SUCCESS in 54m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/9be58071b17b462ea61f9db8cedbe024 : SUCCESS in 50m 28s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b61883fed1854684ad85e2d80aed68a9 : SUCCESS in 39m 05s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7d167607e159449baba27fdb7dc56b99 : FAILURE in 1h 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c73cf61d63674c19656953a48c1d1206c924ae73","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-07-28 20:43:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c5d91246d39a4321ba384e1973236340 : FAILURE in 39m 27s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/286065f7b172491bbb4066fb683f4608 : FAILURE in 38m 59s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/123bf6c7b6484cd0bd295cea8822b9d4 : FAILURE in 38m 30s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7e833ec187744aefa5c89aab3812fcae : FAILURE in 48m 01s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a9e7311fffe24d8baf2d94683fb33b00 : FAILURE in 38m 13s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/914ab7fb6f2343148e6484eaadcbf2cf : FAILURE in 37m 29s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1bd65fa5d4b24732bdc184a63277cb7c : FAILURE in 37m 10s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e599bff874844635ad26b3faccbe5b37 : FAILURE in 36m 04s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"62922985acd3a9417ce7441886e51a88076b7569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-07-29 14:13:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d4e637da0806f904cd10542611f0f25464b160c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-07-29 14:16:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d6e4414555484a9dc90bd425dc87e7ae40404c9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:13:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f83a6640214a7a9a53b2f61198acd6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/022e9bcc496d4f5eb56a5284b893d717 : FAILURE in 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a1c12e83fcf40b38c8493252830924b : FAILURE in 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf076e50cb88728e1a13d2e6f3d323ada96f0e0c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-07-29 15:52:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/df9c0f0/df9c0f0.html : FAILURE in 5m 12s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/4c9336c/4c9336c.html : FAILURE in 6m 03s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/d673579/d673579.html : FAILURE in 6m 41s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/1f48994/1f48994.html : FAILURE in 6m 14s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/f34992d/f34992d.html : FAILURE in 5m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/851413/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/e0c8dfc/e0c8dfc.html : FAILURE in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"976769c1a0e6c70476effae881b7afd4edfd8eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:52:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02dee7dce3ce4144abcb6aacce4a273f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f4d0261f324dbb9fc78f67079e2613 : FAILURE in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78b31dfc2fc49909959e43fe1231cc2 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/985279e869714515a78dff379ff50b7a : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8277516861674fc2992d8e8ace34dc3e : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114264443b344dcab973b47b530a93a3 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f037fff103f4c8f952ed2bdada05e28 : SUCCESS in 15m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85ece930a28f49f2a510a981a67f9298 : SUCCESS in 6m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2c7fdc90fda140e5ba1d57f46132e869 : SUCCESS in 10m 33s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9c69b8891e23497280767ae0df0c9f56 : SUCCESS in 38m 41s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/bfbd123238454697b905bc63447b92c7 : SUCCESS in 1h 04m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/007e7800201b4dec857fa91e4c48bfb7 : SUCCESS in 1h 05m 43s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/01718a5bf0034d82b10489339e071d4f : SUCCESS in 55m 05s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2a6ffbfd54334a7cb7e853f24f91c781 : SUCCESS in 50m 39s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/51c2391a7b0344588582a0c6c2fad97d : SUCCESS in 1h 09m 35s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e5a8594820c14e24a4cffb94703ffedd : SUCCESS in 1h 08m 08s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/4f652cae154c428fa4407c5632dbab14 : FAILURE in 4m 05s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3471df4d448e4e308daf334505f52657 : SUCCESS in 1h 05m 33s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7f3683d2e69a4a4fa56d21820d63848e : SUCCESS in 57m 23s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/13738bebe086483cacb4d9ecde30bf7e : SUCCESS in 47m 47s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/949082e3870b48e1bd7d3b4cef45488b : SUCCESS in 38m 12s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/95888d5b3b8846efbfdc4f05807392a5 : FAILURE in 1h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2699c6d34fea068fc465dd9c04dc5961b2636f1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-07-29 18:34:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7923585ad91048b3ba77bc41b5b6dcb5 : FAILURE in 30m 19s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5b7009e162b440cd939baa2de6ce9d47 : FAILURE in 29m 47s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/56bb8a87f0ea46b196b1926f4d639660 : FAILURE in 29m 47s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c3c30bbb32684ef198236b2afaa669ad : FAILURE in 38m 17s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2a0c3159ce834cebbac869166b93a78b : FAILURE in 37m 59s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b9c9cc4645e740f0b063ca4e7a41d63d : FAILURE in 39m 24s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/47f60c8564d140a6a7810a97be9ec4e9 : FAILURE in 39m 18s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/98214693e89c4bf1a7f00b5a755ccb44 : FAILURE in 41m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"0060a180eccd184a1afa52fab39062078e57e6c3","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-07-31 21:35:05.000000000","message":"Patch Set 2:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with FIPS enabled \u0026 the other is FIPS disabled, both with CentOS 9 Stream\nand PXE boot for Ironic node boot, then we run following Tempest tests in both environments \n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-fips finger://base-openstack/295f294ed1394439845dbf68a9f62864 : POST_FAILURE in 36m 42s\n- fujitsu-irmc-ci-job-nofips http://openstackci.jp.fujitsu.com/irmcci/13/851413/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-nofips/d51751e/ : FAILURE in 36m 24s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f99d6b2af27250776496baf774c74c44e234140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-03 14:33:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"084b22fd1ed6cd011a1e20c1313e09c773d2ffb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 16:10:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a33b4e3ad87428ca483c3434b499d6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b95cbf3521eb45c0b0245f6143b47e9f : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4882971eb04c4c17b259bf9b0c29f983 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2671e325597e45c59a28afba2542e5be : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/927119af71604ad0b9401fc97d8476a9 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b4b4d8b8bd49e8a290bfe6ce439d00 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1830681262744f7595a18e7622da8b65 : SUCCESS in 50m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c3a1e2a380047c992bc045ed2b9b036 : SUCCESS in 56m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ea1bdd1d1f40ecbc3dfdc5fdd5167c : SUCCESS in 17m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc5a11a8432a40528e82d362c4609e3f : SUCCESS in 7m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/62d37910b3d64bbbb152703898071292 : SUCCESS in 11m 17s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/21e1b787ff83469ca5c5313bdb900b46 : SUCCESS in 25m 02s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/ad7f44b7146a43f7b1cd545f31ae6ba2 : SUCCESS in 55m 08s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/47a589b363e34b4a8852a79b490d2738 : SUCCESS in 1h 03m 08s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5f158950bcdd49fa863b108b79d75d1c : SUCCESS in 1h 02m 35s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e2c47fed0d774ae685f4f02a3a2a7ed4 : SUCCESS in 1h 04m 53s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c3f460a202994e9a896151b685559e68 : SUCCESS in 1h 06m 12s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/151a7a00103446cea23e8d72218d5a59 : SUCCESS in 57m 41s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/64db04ed30124ba497d42bbb1fb7e7ac : FAILURE in 5m 00s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7305060599b2442182fa4a0096bad555 : SUCCESS in 1h 04m 56s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3ecb52e911ae4c2bb87d7691b83db7e8 : SUCCESS in 1h 04m 29s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4a1fa6404cc54c4cb8fd25824cf1e352 : SUCCESS in 51m 54s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea9af63fab24892aac9a8d6f52a02ef : SUCCESS in 41m 43s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/596af4a25ded465ea513728b8030f71f : FAILURE in 1h 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"057861d93fb7a96d2a79a79af7e8e32e03438e96","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-03 17:27:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/25d210dcd0484323ac0ef8a643d8716b : FAILURE in 29m 44s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d905a1cdd8fd446ca00ed77129b1e632 : FAILURE in 29m 45s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/18a8b430883742fb9557e86a078353ae : FAILURE in 29m 33s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6af8cb7c4e9d496f830c58bb119787e9 : FAILURE in 30m 34s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f2cac088f2d84ad8a260d6ff4c3cd598 : FAILURE in 30m 25s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7312070787574cd3a3a20b63bcdfa252 : FAILURE in 39m 42s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1ec56c8c146b4dbd84125dbaffd92a0f : FAILURE in 30m 08s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2dcc6e0be8ae483d96048d481cab6a88 : FAILURE in 29m 48s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d2f4e21dd6f642b8a90c3c82de23af81 : FAILURE in 30m 04s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e6407896ec81479c89f5964ecf3919bb : FAILURE in 38m 50s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/632aa62dfecf42269167e485647a1b84 : FAILURE in 38m 57s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"75f32bf928ec687ea00479d03884f122ac867ef9","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-08-03 21:23:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6c974c71258fa50ae2e0a505be1612f4f45983b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-04 08:34:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"483d213a012ecb54283f58895fafff39196d8b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 08:35:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d15b5196938d4d18b70a2197ae538b6323615a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:50:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0cc1d2d90d465aa65498d0e8356f22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/347948f3a56040d4b2924502565ef2f3 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21caf2f3fcf47a0814b1d6c0c83c303 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380ba577e1f54f129c0a619e131cacc8 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83cdf3f687e240fab208f086a61be2a9 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4857449222d44967999cebbb08f60bc9 : SUCCESS in 14m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37d499c95db42a4b706aa17ae937040 : SUCCESS in 7m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5ed6e4dcbc1e4c278902f65a50b529c7 : SUCCESS in 10m 28s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/20c8c0b0c7b44e72a1f7f29ddefd5bdc : SUCCESS in 33m 58s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/fa09d9fbcb4c4fda844e1d7f4874dbfc : SUCCESS in 1h 05m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e67e1a63e44d4384ad79c22bcb046f55 : SUCCESS in 1h 12m 33s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3902c237d8f94053b8b27648befc70d8 : SUCCESS in 1h 01m 29s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/dc3e82a890e046079890040bc584c0d0 : SUCCESS in 1h 01m 57s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/eea3153d85d741fb963a5dc19366fe50 : SUCCESS in 1h 13m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/fdd0b378b79549b2a647e4395e183271 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":3},{"id":"bd0841d55be1440f378f5f72729680eb4d676410","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:50:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8bb9da3470ed924813bf6a2b2af5eee5f53ee787","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:51:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e8a9c64bee468e82ef1f1b1d5502f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee8b92496b84c9d8d45087d7d1b78f4 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8b66b7d207843849bc2f5161b269484 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"046ddf4e6a1260e96227cfe39c0ad2d29923d621","revisions":{"ddd540141d4b6253048f17439befbee84dc47774":{"kind":"REWORK","_number":1,"created":"2022-07-28 14:18:26.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/13/851413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/851413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/851413/1"}}},"commit":{"parents":[{"commit":"c5d938a6a9e32eca9a3cea99494207c5ddc97e09","subject":"Merge \"CI: Removing ironic job queue\" into bugfix/18.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c5d938a6a9e32eca9a3cea99494207c5ddc97e09"}]}],"author":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-12-28 12:16:01.000000000","tz":-300},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-07-28 14:17:37.000000000","tz":120},"subject":"Fix Redfish RAID for non-immediate controllers","message":"Fix Redfish RAID for non-immediate controllers\n\nFixes ``redfish`` RAID interface workflow to not create more than 1\nlogical disks per controller at the same time when controller does not\nsupport \u0027Immediate\u0027 application time and system needs rebooting.\n\nWith this fix virtual disks are grouped by controller and executed\nin batches when reboot is required.\n\nThis enables ``redfish`` RAID interface to be used with iDRAC systems\ndirectly as although there are many controllers that support \u0027Immediate\u0027\napplication time, they do it only when system has completed loading.\nSystem intermittently completes loading after IPA starts executing\nclean or deploy steps rendering controllers not supporting \u0027Immediate\u0027\napply time and requiring reboot. This fix handles such scenario\ninstead of failing to create more than 1 virtual disk.\n\nStory: 2009863\nTask: 44529\n\nConflicts:\n       ironic/drivers/modules/redfish/raid.py\n       ironic/tests/unit/drivers/modules/redfish/test_raid.py\n\nChange-Id: Ia2ce34f09695731b0f48798f662006c4904e2223\n(cherry picked from commit d7d8f8754b36af4b686ce540e699fc5cb945c10c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ddd540141d4b6253048f17439befbee84dc47774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ddd540141d4b6253048f17439befbee84dc47774"}]},"branch":"refs/heads/bugfix/18.1"},"b3861d02146835a12901f4d244ec036d5faee855":{"kind":"REWORK","_number":2,"created":"2022-07-29 14:13:54.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/13/851413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/851413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/851413/2"}}},"commit":{"parents":[{"commit":"0a84944e7d475873075f2d2f072fb8f64993c6e2","subject":"Fix redfish RAID failed tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0a84944e7d475873075f2d2f072fb8f64993c6e2"}]}],"author":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-12-28 12:16:01.000000000","tz":-300},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2022-07-29 13:45:45.000000000","tz":-240},"subject":"Fix Redfish RAID for non-immediate controllers","message":"Fix Redfish RAID for non-immediate controllers\n\nFixes ``redfish`` RAID interface workflow to not create more than 1\nlogical disks per controller at the same time when controller does not\nsupport \u0027Immediate\u0027 application time and system needs rebooting.\n\nWith this fix virtual disks are grouped by controller and executed\nin batches when reboot is required.\n\nThis enables ``redfish`` RAID interface to be used with iDRAC systems\ndirectly as although there are many controllers that support \u0027Immediate\u0027\napplication time, they do it only when system has completed loading.\nSystem intermittently completes loading after IPA starts executing\nclean or deploy steps rendering controllers not supporting \u0027Immediate\u0027\napply time and requiring reboot. This fix handles such scenario\ninstead of failing to create more than 1 virtual disk.\n\nStory: 2009863\nTask: 44529\n\n Conflicts:\n       ironic/drivers/modules/redfish/raid.py\n\nChange-Id: Ia2ce34f09695731b0f48798f662006c4904e2223\n(cherry picked from commit d7d8f8754b36af4b686ce540e699fc5cb945c10c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b3861d02146835a12901f4d244ec036d5faee855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b3861d02146835a12901f4d244ec036d5faee855"}]},"branch":"refs/heads/bugfix/18.1"},"046ddf4e6a1260e96227cfe39c0ad2d29923d621":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-03 14:33:01.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/13/851413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/851413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/851413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/851413/3"}}},"commit":{"parents":[{"commit":"ed0b22f882e97f1fcd0d8af899d0cbc79b68f207","subject":"Fix redfish RAID failed tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ed0b22f882e97f1fcd0d8af899d0cbc79b68f207"}]}],"author":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-12-28 12:16:01.000000000","tz":-300},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2022-08-03 14:27:42.000000000","tz":-240},"subject":"Fix Redfish RAID for non-immediate controllers","message":"Fix Redfish RAID for non-immediate controllers\n\nFixes ``redfish`` RAID interface workflow to not create more than 1\nlogical disks per controller at the same time when controller does not\nsupport \u0027Immediate\u0027 application time and system needs rebooting.\n\nWith this fix virtual disks are grouped by controller and executed\nin batches when reboot is required.\n\nThis enables ``redfish`` RAID interface to be used with iDRAC systems\ndirectly as although there are many controllers that support \u0027Immediate\u0027\napplication time, they do it only when system has completed loading.\nSystem intermittently completes loading after IPA starts executing\nclean or deploy steps rendering controllers not supporting \u0027Immediate\u0027\napply time and requiring reboot. This fix handles such scenario\ninstead of failing to create more than 1 virtual disk.\n\nStory: 2009863\nTask: 44529\n\n Conflicts:\n       ironic/drivers/modules/redfish/raid.py\n\nChange-Id: Ia2ce34f09695731b0f48798f662006c4904e2223\n(cherry picked from commit d7d8f8754b36af4b686ce540e699fc5cb945c10c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/046ddf4e6a1260e96227cfe39c0ad2d29923d621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/046ddf4e6a1260e96227cfe39c0ad2d29923d621"}]},"branch":"refs/heads/bugfix/18.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
