)]}'
{"id":"openstack%2Fironic~851900","triplet_id":"openstack%2Fironic~bugfix%2F19.0~If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8","project":"openstack/ironic","branch":"bugfix/19.0","topic":"uc-yoga","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-08-03 12:44:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8","subject":"Use yoga upper-constraints for all tests","status":"MERGED","created":"2022-08-02 15:24:57.000000000","updated":"2022-08-03 13:02:17.000000000","submitted":"2022-08-03 12:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851900-uc-yoga","meta_rev_id":"dc2f389618f19af30ef14a40d61f0cadad0c0f12","_number":851900,"virtual_id_number":851900,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-03 12:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-03 11:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-03 11:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 15:29:37.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-02 15:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 15:30:20.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-03 11:25:44.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"1196ec5bfcb4a5c686719fc6a0f9e975db7b862d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-02 15:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d03c6620b3d17373ad051f6905beabdb7cf5f1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-02 15:29:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4917ac33fcb05131a1fd88ff3096fbd288fc39c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-02 15:29:37.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch bugfix/19.0.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@bugfix/19.0\", line 1141, column 30\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"de4e6c81b0f92eaf5bb826e84cc19171071b2952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:29:52.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch bugfix/19.0.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@bugfix/19.0\", line 1141, column 30","accounts_in_message":[],"_revision_number":2},{"id":"6e1ffa1f177c394b5a8068a198aaa03e8b4c64a0","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-02 15:30:20.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic on branch bugfix/19.0.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/ironic/zuul.d/ironic-jobs.yaml@bugfix/19.0\", line 1141, column 30","accounts_in_message":[],"_revision_number":2},{"id":"e47626d027fcba7707255e25c7fa07546ded05cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-02 15:46:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec2556a5231779293beb792630ae0ac22dec6b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 16:36:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6799f6e6526744d3868635921b39458a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e80ae7f8b923401c84b39049e965b03f : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c105fba42374432084063add1e9c590b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33b2c267169d4ec1960d6afdab52817b : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70ba79802948430aa6854633a448852b : SUCCESS in 10m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce88504956044e52b1fa43794a57eb9a : SUCCESS in 18m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e82a4f0ae15848e2a4e359fafa74d84d : SUCCESS in 15m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e9c49d17ee4522b86b1f291beaeaa2 : SUCCESS in 15m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/27eff13248424fef87fb9e73adb716d0 : SUCCESS in 9m 40s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e4c88aa4988f47b9891a1278d9546280 : FAILURE in 10m 21s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f077188aa2c74038974dca101b4aeca0 : FAILURE in 25m 48s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/6f45da0315d14903b71701a1c751bf42 : FAILURE in 24m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/22215c6e905a4fcab4192db3979e370e : FAILURE in 28m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/49cdbf87381141d782aa28bb4b892b50 : FAILURE in 29m 09s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ed285a7a476a40d0ab22c9dec13d59c1 : FAILURE in 31m 34s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/31f9e4058da9418b81b3167d6c0109d3 : FAILURE in 38m 06s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ee9bdc726d154dcf8446f1eb299371e1 : FAILURE in 30m 36s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/b2934754d5a44bb19c8eedefc3a3f69f : FAILURE in 39m 10s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/c3c9798c24424be7a123baf5fe76da19 : FAILURE in 36m 34s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8fe83105f2d841c8894afb9d6884a840 : FAILURE in 40m 43s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/204b2c6029604cb0a8ec33d17e690e3a : FAILURE in 41m 24s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/853e2507a5ef4e5a996219d095712ac5 : FAILURE in 38m 37s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e1fbed550fe849279528c47085f31a5a : FAILURE in 39m 15s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/8cba947bcd144edc9788db773d040c05 : FAILURE in 3m 41s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/bf6629983b484aa1b9e41d9c076ff22e : FAILURE in 42m 35s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/12f8b2dceb724b4b8503a939de1ff10e : FAILURE in 42m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/9d5e71897a53429198af1b5626e591c0 : FAILURE in 31m 36s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5d78e5b93a0c451c9669a645157465f8 : FAILURE in 40m 00s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce593fd6dc7c421d87d5c9ecdce70fea : SUCCESS in 37m 52s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7d5096fdc7a7422ab64ca28bb8fb56c9 : FAILURE in 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0879ed4687a72f54a2f8bb2aac46894e65ad974","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-02 17:00:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/53f3cc2b62164e079e2fe964a6da015e : FAILURE in 36m 56s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a74fe1198d2a4814ba3c66df0a08b186 : FAILURE in 36m 09s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/800fe194a14943458c38c459a8515335 : FAILURE in 36m 19s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/88a1bba767d4470a85c10a0390d6d17b : FAILURE in 37m 13s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/30a8550e61e544f2824f2d3575e4c702 : FAILURE in 36m 05s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/38e0cf8463824b529ea98f5e3faa808b : FAILURE in 36m 15s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5d6646978b3d475496d64ac995309f80 : FAILURE in 35m 23s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/576e28b89c7c4885985fc4b6bce9ed7b : FAILURE in 27m 01s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/259a4035569348c0a3a7133de7d00876 : FAILURE in 35m 01s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ccb643f0403c4be2859bc6a523b16508 : FAILURE in 35m 15s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7779026c41ae49109f51cc7a0a06d32a : FAILURE in 35m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"0de243348e52d565c7b566302c7866fdf816dc02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-03 07:11:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"47883d101cd1a003ce5491299efc59c87cbf653c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-03 07:50:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c0510125c9ef17988a84dea13b395d22d62eba66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 08:42:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f0c2af6d6a548848ab8b7e6876200da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7282947f58a54050b6ec38cd1308fb29 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2e6a998ced43a482372dd10ae4d7b4 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/152d15cebb1f46ddb6dfe3418834a41d : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8886c344f8564f8785f3567ab047f944 : SUCCESS in 8m 29s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f80316c5288d46b58e894e49ff960e70 : SUCCESS in 16m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c93fef163a8c45f2ae6de1d320705e13 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2682f0f65d104b1d85d27948f54b1501 : SUCCESS in 13m 34s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/91ad5d4f0571406eb869d02ac768875c : SUCCESS in 8m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6c218fa189fe477786d1d68e045b2a89 : SUCCESS in 8m 17s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f03ba7ac3cf1422eaaa615b91f02bc8e : FAILURE in 32m 40s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0ebd1bccd86c4987999d9d7e634b3246 : FAILURE in 21m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/82209e61fa5e4a0f9007002648515ccf : FAILURE in 35m 20s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/89ffc3e0d9384782a852a75e52adbe4d : FAILURE in 36m 33s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9a2e1411cade4a4eb6972be9143bd0e7 : FAILURE in 37m 37s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/eb0d21a990574cfd9cfbd91e3c24837e : FAILURE in 27m 52s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8dffa26e5b424ea6bdb727d46acea1c2 : FAILURE in 39m 22s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/7ee73e367eb2409f960d8f16ccb5036f : FAILURE in 26m 01s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0ed3693af4b14eebafe235e1b7fc8d8a : FAILURE in 40m 27s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/07561e8fb1f2496fb8d011d3fab2efce : FAILURE in 39m 09s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/00a2a60a0f594671a0397d10236e6617 : FAILURE in 27m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b0d2c761cc2a4076abe821f1632e2f64 : FAILURE in 42m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a6aa4e6c83bb438b9b4aefa9c1aeb3be : FAILURE in 24m 36s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/53f911f310164861926317bc07d135a8 : FAILURE in 4m 21s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7a89ce58110a45ad938fa58aa4d3fe65 : FAILURE in 44m 02s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9e6873cf318946e59357c19cc32ec5a9 : FAILURE in 39m 52s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4b830e541f284992a62e99d6ce342f1c : FAILURE in 27m 45s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1052bd5dec854e588ed68abca6b7bba8 : FAILURE in 41m 07s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8682685044f84e83a351ce1d5adb7ad0 : SUCCESS in 34m 11s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/90339cd7339045538b1d3eb99fcdfe81 : FAILURE in 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"180cd78ffd1c2e431d4949f21fe1bfefbfb22b54","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-03 09:01:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ea3e077f42d84b54ab0f84dd6a2fa922 : FAILURE in 27m 30s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2a761e6dc3eb4470b0b2d7c076771a4c : FAILURE in 27m 52s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5a4cfa27666a4ecbaedc5a6753d4a1a2 : FAILURE in 27m 51s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e0cccb32a3f94af7b0e56b602f543346 : FAILURE in 36m 47s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/72f1249a29f843d2a2ce3f51fa6e9ccb : FAILURE in 27m 44s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9450c1d36ce143d38fa050d8aea16972 : FAILURE in 37m 26s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/dba49f20228f4365886eedddf5b1c833 : FAILURE in 37m 39s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8224967467df42248ce6333a3333882b : FAILURE in 28m 28s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8ba421de216243dd9d43c3a62dd15ab8 : FAILURE in 36m 40s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f2ca24605650485183b944dcfbbbb932 : FAILURE in 37m 25s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5bbfbb90fec7426fa3309e2e5bbe4e31 : FAILURE in 27m 56s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"36765b15d9bc92d68eff31b2a1df3689f4515fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-03 09:22:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"84814483277dcbbe76daa11531e87ceda94f06dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-03 09:26:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fe0f613a4531cc28b7d800c33fa4d518cd613917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:04:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2693914e29d4cf1b87c9451535ce9db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248c987b429742ce983c52a829d2ec41 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4e9af50cf1493f9a648cbcc75041d8 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fd9c74d7740430eb49c798e8e95f4f1 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b6b22bf8a64f1c84f944d2bbdf97a1 : SUCCESS in 8m 53s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7498708654544df8e8ffacc051bbbf1 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/173852f776dd4d0d9aa07007855aa3ab : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d5537a685c4cfe8023f335d362c7e6 : SUCCESS in 12m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/aab550f2b12f47f796cf079160ae025a : SUCCESS in 10m 25s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f385d8673e4f41c9909ee750efad0bbe : SUCCESS in 7m 53s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ae95162fd591464a96ec4db36193e3d8 : SUCCESS in 34m 00s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c48a28b9f493408baccc7575f729a8bb : SUCCESS in 32m 03s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/cf7349cd709d4ed6a1028889d061380c : SUCCESS in 1h 14m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/afbf393a6b3144708cfbe227b277d4fb : SUCCESS in 57m 06s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1e16534487a04a74904bb991653478cd : SUCCESS in 56m 49s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0cea0c87e4414bd18317976e70bff865 : SUCCESS in 59m 24s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/821224dfadb341b3b71f2f7c27708b19 : SUCCESS in 50m 38s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/d9290570cbbc43e0ab5d26377c8e2523 : SUCCESS in 1h 10m 42s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/a137d267ae1a4a01a63f0a07fe80e205 : SUCCESS in 57m 53s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/7343c0d38f634418bd2899afb324180b : FAILURE in 46m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/1f5a71fad1574f3bb477a675a429fb98 : SUCCESS in 1h 07m 26s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/77431a48f15c4c51a0585461dd5446f4 : SUCCESS in 44m 07s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3dfefb1a856547b0967fae19e7a59dc2 : SUCCESS in 55m 29s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0208c267782a4ac7aa02514f387a0aa7 : FAILURE in 5m 05s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1843c5bedb154f39a05d41725af2822e : SUCCESS in 1h 03m 09s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/93525dd2135a4644bcc4c8777c5d461c : SUCCESS in 56m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/faac04764e65429b8e5b8764f7fe66fb : SUCCESS in 55m 24s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0e8d2881dec44b4b83e2c6ce32315e61 : SUCCESS in 1h 02m 17s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae9f02749f14681a63110f978f9d6c1 : SUCCESS in 38m 25s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/0f1084d6f29c45a7836c2ca847f5aff2 : FAILURE in 1h 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"875712e03728e6a5a60352345e1c9fdbc8b0e65d","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-03 11:25:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e6d18b0176b6e060790e272d7ee9ccd065ef21ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:26:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4ab1c7305b86657547ad39a2b281284c05f86b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:44:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90ae419511d4a47807dedd05b09b67c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/646f998183ed4397bb12e9fbd5baf7d6 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79ef11ac4c742489783897d8d6013b2 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe9c07ca46640cca4ce5c0f3b741e54 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fe55530667947d296ffeec493be7ef3 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee2aa58d00e485d982d150030e564d5 : SUCCESS in 14m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ceb438e8181e4b32bafe27e60322989d : SUCCESS in 9m 33s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cf207469682c4a84a9f0ec1d9e10237c : SUCCESS in 27m 53s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/29311cb01fe046048c024cefe4750dcb : SUCCESS in 28m 48s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b79d27182d524eeb8f4992ea33af30dc : SUCCESS in 1h 04m 03s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c7824b8134944612809cb8a4fbd79486 : SUCCESS in 1h 01m 51s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1348c88e73314319b6364facdffa3800 : SUCCESS in 53m 06s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4a915ac1b2b24d2e9c89f4bfe8456892 : SUCCESS in 50m 49s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4fd08d3decc74711b37b1f093710ff6e : SUCCESS in 43m 14s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/0a8c193007d4474593392ebe871c4f5f : SUCCESS in 1h 13m 16s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/af33f6b6860c4da8b3fd2c383570bbc1 : SUCCESS in 53m 09s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f9fb45b5ee914657acb6f1d7c5c3e76f : SUCCESS in 1h 04m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e6d22c614f5542d0bf3a08fd6d490ca0 : SUCCESS in 38m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/e78e4b30ad024e30807f3b0fe1fc7342 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":7},{"id":"bad4ed76d0e309c67a748668bddd218c83586178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:44:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"85877be38ad578217a65c56d4a62b045b2cd5eea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:46:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91a98ab7b84d4244baee8309cf3ce14b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5900ac41e6f45f9a807b9f67ddacd90 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":7},{"id":"dc2f389618f19af30ef14a40d61f0cadad0c0f12","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-03 13:02:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/22d1f44756704fd898ed4d9f1737ecef : FAILURE in 34m 46s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2d6543e283694939987a2a32f604cb40 : FAILURE in 25m 59s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/786d55e9505448a4bdd8fcd684f5bb1f : FAILURE in 34m 06s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2e5ee2d8d51f41849cf95ec0d7bffcc5 : FAILURE in 35m 02s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d1441608bbde409098e5fb07bba6993e : FAILURE in 34m 15s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fb7df755e28f4897b30dd4db1b99d7d0 : FAILURE in 32m 38s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e003e4eccf274c3aa65602bf3a5b33b6 : FAILURE in 34m 16s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6516e4638ee2485a9db110af8226e74c : FAILURE in 34m 29s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/05f8a27a31924a9da28611a27266b891 : FAILURE in 33m 57s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b90c9df5d9e84b60a7a5466d35ad7894 : FAILURE in 32m 28s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/dbb7f3b2531743478acdd1254771b122 : FAILURE in 30m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7a20175549ab6f21cc148a03bc17c0a26faae009","revisions":{"0b44f6d418df3345cf5badca37422df7645dbcf3":{"kind":"REWORK","_number":1,"created":"2022-08-02 15:24:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/1"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0b44f6d418df3345cf5badca37422df7645dbcf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0b44f6d418df3345cf5badca37422df7645dbcf3"}]},"branch":"refs/heads/bugfix/19.0"},"fa3414a1369203f5d8f2dc2192938c0fc2b6ab23":{"kind":"REWORK","_number":2,"created":"2022-08-02 15:29:20.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/2"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:29:14.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fa3414a1369203f5d8f2dc2192938c0fc2b6ab23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fa3414a1369203f5d8f2dc2192938c0fc2b6ab23"}]},"branch":"refs/heads/bugfix/19.0"},"e1aa60ee9174472b2f45c482d60e657fd6347a63":{"kind":"REWORK","_number":3,"created":"2022-08-02 15:46:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/3"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:46:08.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e1aa60ee9174472b2f45c482d60e657fd6347a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e1aa60ee9174472b2f45c482d60e657fd6347a63"}]},"branch":"refs/heads/bugfix/19.0"},"3d9b8d376327bf1660e042e7da66292e3fcae43a":{"kind":"REWORK","_number":4,"created":"2022-08-03 07:11:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/4"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-03 07:11:45.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d9b8d376327bf1660e042e7da66292e3fcae43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3d9b8d376327bf1660e042e7da66292e3fcae43a"}]},"branch":"refs/heads/bugfix/19.0"},"88c1f4c42a1a97066f54ce6e19fd370b9f3363bc":{"kind":"REWORK","_number":5,"created":"2022-08-03 07:50:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/5"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-03 07:50:51.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/88c1f4c42a1a97066f54ce6e19fd370b9f3363bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/88c1f4c42a1a97066f54ce6e19fd370b9f3363bc"}]},"branch":"refs/heads/bugfix/19.0"},"5de619c77a75fb355d3845b7327685e82df42b08":{"kind":"REWORK","_number":6,"created":"2022-08-03 09:22:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/6"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-03 09:22:51.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5de619c77a75fb355d3845b7327685e82df42b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5de619c77a75fb355d3845b7327685e82df42b08"}]},"branch":"refs/heads/bugfix/19.0"},"7a20175549ab6f21cc148a03bc17c0a26faae009":{"kind":"REWORK","_number":7,"created":"2022-08-03 09:26:25.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/851900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/00/851900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/00/851900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/00/851900/7"}}},"commit":{"parents":[{"commit":"f14c48811ee865b58104a43ea30372ef9e383fca","subject":"No deploy_kernel/ramdisk with the ramdisk deploy and no cleaning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f14c48811ee865b58104a43ea30372ef9e383fca"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-02 15:24:40.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-08-03 09:26:20.000000000","tz":120},"subject":"Use yoga upper-constraints for all tests","message":"Use yoga upper-constraints for all tests\n\nChange-Id: If0d08fc151afbdba8c7f65ae1fe8b1f6aa949af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a20175549ab6f21cc148a03bc17c0a26faae009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7a20175549ab6f21cc148a03bc17c0a26faae009"}]},"branch":"refs/heads/bugfix/19.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
