)]}'
{"id":"openstack%2Fironic~852639","triplet_id":"openstack%2Fironic~stable%2Fyoga~Id66b87309dd26a2a165b35ac1d81580e4605d629","project":"openstack/ironic","branch":"stable/yoga","topic":"redfish-raid-stable/yoga","attention_set":{},"removed_from_attention_set":{"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2022-08-11 18:31:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id66b87309dd26a2a165b35ac1d81580e4605d629","subject":"Update raid_type handling for Redfish raid_config","status":"MERGED","created":"2022-08-10 09:47:22.000000000","updated":"2022-08-11 18:33:26.000000000","submitted":"2022-08-11 18:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852639-redfish-raid-stable/yoga","cherry_pick_of_change":852393,"cherry_pick_of_patch_set":1,"meta_rev_id":"f295399439838246e56f8273b83327775a3e03a8","_number":852639,"virtual_id_number":852639,"owner":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 18:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-08-11 16:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-10 10:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2022-08-11 16:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 10:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-10 10:16:37.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-08-10 11:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 17:30:10.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-11 16:37:05.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"63fd01ce80ce8c3218af791a4226458db97e24c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-10 09:47:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e4afa8c070ac1221be49542aa06fd696e8f4a61b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:11:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1418f83cac02426d87e09b5e6dd5f843\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34b87120ab3e4d1da7a1cec49f257aee : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdd00bac5ae7455abc33af504e9d437e : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fe9382ff92c42ed4871497494e3c41ba1d18f88","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-10 10:16:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"addbe00d556a924c9fa27b7d2ad033d91b672186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:42:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae224d4fa591488e850f05c3017195cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0fc025b6407483aba25b043a9a0e867 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbd6f562cc343a7ab51354662fdf971 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0472d993c6584602918738f588fe3375 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/801aa7e7df28444596ab13eb8bb127ae : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1ff3784a9e46ccb2a0a0f4c289eb24 : SUCCESS in 11m 48s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/53eb7c0c67cf4b37b31f1c4ab0ba5135 : SUCCESS in 11m 00s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e68d5771d6674a3f82b366beb3144f49 : SUCCESS in 7m 46s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bfd4d3a1b508486eb2eee83c544f3fea : SUCCESS in 25m 54s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a2fccbc52d774f6fa4642080eb07819d : SUCCESS in 29m 12s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/59762049eb5f4b25874cf7aa83cd4e79 : SUCCESS in 1h 25m 35s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/6ff8369ed757424881bf2f1de741f55e : SUCCESS in 52m 40s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f335dd1385314e4bbe46fb390f1645eb : SUCCESS in 55m 01s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/af426f413bea4feba55a1ff79dbbfbcb : SUCCESS in 49m 16s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b486cdb1b8d14ff587268d5ca971534e : SUCCESS in 1h 05m 51s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d211aa132ba2494787d05adb9abfb930 : SUCCESS in 42m 43s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/10aefab9bee24cbbbb8bb6a4a3186beb : SUCCESS in 1h 24m 01s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/99ebf46e856b48199d586da5d3801816 : SUCCESS in 1h 07m 49s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/a15da2e549394bebbe10bc287edf5eae : SUCCESS in 1h 52m 43s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/64b969acfb984ecfb8e4e93ea5da6093 : SUCCESS in 55m 07s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/996895a4da8f4a67a269746304a9634c : SUCCESS in 35m 30s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/76560635d789478fa03f18f9e6197ba6 : SUCCESS in 48m 25s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/d3d7a9b1786547bebd364ef15d43a509 : SUCCESS in 1h 43m 07s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/e6ea6738fb434a61950c1d2c1bf1c1c7 : SUCCESS in 4m 27s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e5b9432ddf424c6b9257ba95f5444ae1 : SUCCESS in 54m 06s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/4de7e126c4ef48c39ba83d9fde9f07b8 : FAILURE in 59m 59s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e9caefe3cc624a369b6189ae730acc65 : SUCCESS in 1h 04m 34s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f5ff6aa114684ad09023f961515a1d25 : SUCCESS in 41m 27s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/aabcb02e6c0546ee92b74ea2fbe351e3 : SUCCESS in 53m 10s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62ba2303d27c44e0a7b232bcc0aa9cb3 : SUCCESS in 30m 03s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/fd5b1524bc2f412696ed5369c8c41b73 : SUCCESS in 28m 22s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b6128dbc6b0b4dfa9064818611466777 : FAILURE in 1h 20m 00s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/459a840b9f0a47b1a6b829fd944441b8 : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bdd8d827fa782e25b19e88f127cba831a8f9068","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-10 17:30:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/91181f191c76429782721c30aca106ed : SUCCESS in 55m 35s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7e3f4ff05faf4fd3b7d30edf8eb1d1da : SUCCESS in 52m 18s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d550681a96664823a9966eef275c5407 : SUCCESS in 52m 55s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cd9fe9ffd75f46eb9b6dc2d187a9de97 : SUCCESS in 54m 31s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4f8fbdef02c046829161ffb72b280bb2 : SUCCESS in 51m 22s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ac62974a58c34bbba9133fbe90e606e6 : SUCCESS in 52m 59s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/58db77bcaf6b4e76a15adcc973eacb51 : SUCCESS in 50m 34s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cd5a3cf476874cea931498332e0c6dc8 : FAILURE in 27m 01s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c09abca3f2a74546a492b1ed5c8b6d78 : SUCCESS in 54m 45s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/86c243411c8346d1863d2323a6b7ac95 : FAILURE in 27m 55s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/34d27bab6495423c8e981d4fc91389d3 : SUCCESS in 1h 00m 05s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e6f2037b1c2539e012dfcd70c3c825ef30f4f5f","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-08-11 16:37:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abe34b536e31b25858bcc74c7b5d5d1ee509815f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:37:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4930d4d1c587563374834daffe2bb29f26c2839c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:31:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1befd7dbbe4b3bb5c119f842f69c7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a37406862d4944a086c919effd7059 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b21ce64c8e843b3aa0393e465808434 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48a562b4164d4eaa868d0edaa2ba95f6 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c2ec0a5f3f4a8fa2bc24e6ae281413 : SUCCESS in 10m 45s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/27fe97e695cb4db8a3d4bbd968dc8173 : SUCCESS in 11m 49s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/455633df9e1d4ceeb3a7fefe1d514a2c : SUCCESS in 25m 35s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/bc954910b6f14580afb73cab5d493501 : SUCCESS in 28m 17s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/acfc033c44e44230b450e9a3821aa426 : SUCCESS in 1h 22m 20s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1ea540ef9ce04ecfaa85abf5f3cc1227 : SUCCESS in 1h 11m 36s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/69137f9ab32644a18d335e1aa2252981 : SUCCESS in 1h 11m 56s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c94affb6a26c41f79c377c003cdbdeba : SUCCESS in 53m 54s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0162cfbd872d41aaa087455b2eb4a402 : SUCCESS in 1h 01m 04s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/7ff5f877e1f04611b3f953c3c8c2d89b : SUCCESS in 52m 23s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/29a77aea457f4fa28c228bd124819135 : SUCCESS in 1h 48m 40s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3789091f72b24b748550495e2c88eb77 : SUCCESS in 1h 03m 00s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/99b50df1ef72453a90d107e7b7c3c3cc : SUCCESS in 1h 09m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/1ba7da8f989b4e33a407abd32e42aa3e : SUCCESS in 45m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9b1588cd2e61431aa36afea6e0810bed : SUCCESS in 1h 02m 40s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/7534544b694b418ebd858a5dba1085c0 : SUCCESS in 1h 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0e704311935a4494fc751f0fcd6281965f287977","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f295399439838246e56f8273b83327775a3e03a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:33:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad8dc315323407e83d54d1ca8515d76\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59f990811b24b1b8333ca46edc80093 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ee9b67a56805dd00969473335d487173a1bb853","revisions":{"4ee9b67a56805dd00969473335d487173a1bb853":{"kind":"REWORK","_number":1,"created":"2022-08-10 09:47:22.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/39/852639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/39/852639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/852639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/852639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/39/852639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/39/852639/1"}}},"commit":{"parents":[{"commit":"593a547a8dbbbeee2225520812f8b235c39656b0","subject":"Modify test code to avoid CONF modification affection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/593a547a8dbbbeee2225520812f8b235c39656b0"}]}],"author":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2022-08-08 10:15:35.000000000","tz":-240},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2022-08-10 09:47:22.000000000","tz":0},"subject":"Update raid_type handling for Redfish raid_config","message":"Update raid_type handling for Redfish raid_config\n\nFix indentation for case when raid_type is missing.\nGenerally, it is not expected that raid_type will be\nmissing, this is done as a precaution as raid_type\nwas introduced in Redfish 1.3.1. Now log warning\nthat raid_type is missing, thus cannot update\nraid_config correctly.\n\nFollowup to I753c4b00c0a64bcdc89c9bc0afd46f1211f7847b\n\nChange-Id: Id66b87309dd26a2a165b35ac1d81580e4605d629\n(cherry picked from commit 3a621e3983eb2eb1d013d687acc3f5981cdfbc64)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4ee9b67a56805dd00969473335d487173a1bb853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4ee9b67a56805dd00969473335d487173a1bb853"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
