)]}'
{"id":"openstack%2Fironic~854549","triplet_id":"openstack%2Fironic~bugfix%2F21.0~I170d78bdb7ed0f6c36a80a9f2ceb9629f44394ed","project":"openstack/ironic","branch":"bugfix/21.0","topic":"ValueDisplayName-bugfix/21.0","attention_set":{},"removed_from_attention_set":{"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-08-29 10:16:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I170d78bdb7ed0f6c36a80a9f2ceb9629f44394ed","subject":"redfish: fixes usage of ValueDisplayName","status":"MERGED","created":"2022-08-26 12:45:33.000000000","updated":"2022-08-29 10:21:14.000000000","submitted":"2022-08-29 10:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854549-ValueDisplayName-bugfix/21.0","cherry_pick_of_change":854760,"cherry_pick_of_patch_set":1,"meta_rev_id":"e2e2f567e213105721e90b2f03908fb298acb3c9","_number":854549,"virtual_id_number":854549,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-29 10:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-29 09:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-29 09:57:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 12:45:33.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2022-08-26 13:02:26.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-26 13:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 14:54:05.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-29 09:57:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"fd4a1105fc65509a6ef8c369d5dfd9a34f01c6b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-26 12:45:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ebf0d60fe89f8d22a633e6ef01002e5909b5b96","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-26 13:02:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/bdd24bc/bdd24bc.html : FAILURE in 5m 53s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/c97af27/c97af27.html : FAILURE in 5m 08s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/c8a5355/c8a5355.html : FAILURE in 5m 01s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/2688bc6/2688bc6.html : FAILURE in 5m 21s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/227da6a/227da6a.html : FAILURE in 5m 19s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/854549/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/8bb6993/8bb6993.html : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c18c4905bfcefdd3db192d72ea68a51c66a9c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 13:08:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf2c89c2d5149868dc2385cda3ebb2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8be93ed3be3d41c095caf89c5367f2de : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e19f73c1a44036844d3b1703f01b87 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2faf9b4ebffa482185e83ea909e526e3 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd500b08d1a742fa89c13a7bc564ef29 : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66f323231fb148e9b5f8ff848505e332 : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7c23210db9401ca74a7f2eede370f7 : SUCCESS in 18m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/935a24ffb90e4838922b496f5804133d : SUCCESS in 7m 14s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/276767a02d874957b87b841b2f04d7bb : SUCCESS in 13m 35s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2879551327d942369a314a371253ece0 : SUCCESS in 9m 13s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/73ef4e1cea59486498f22bb08457d066 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ae344dac9082db31c59af57472a1cd4dae5e3c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 13:44:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9750844e79a4072bc6a97f2290f016b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a93a07c886bd40b9b469307fa3cf4e91 : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af4acaf2f0914764a803db23d5b6e9b6 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48e1fa1d059f46c612fd71b19b937cfd89f108d2","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-26 14:54:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8626a8c3e2a44ed3b832717d52fd76bd : SUCCESS in 59m 13s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/eadf8e9e7ded4474b14047e55097ce2e : SUCCESS in 51m 46s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d9ca9642cfbf44878fda6cf9946abcc9 : SUCCESS in 50m 37s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/189b9aaceafa446ead0749c736558335 : SUCCESS in 51m 53s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bb4b992ce1874b38b0cdf134668de1b2 : SUCCESS in 51m 57s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e2f72b67d88a4d349c8052ba1d102e41 : SUCCESS in 52m 43s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/eab1ed620cec4403a2bcb37087214c3c : SUCCESS in 50m 25s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f61ac3a80837465ebde503292c5eabef : SUCCESS in 50m 46s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/eb95a91cf44948e1bfc32892b9769418 : SUCCESS in 51m 35s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3b1ede637109472396198d6b4976e667 : SUCCESS in 49m 49s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8c095d3c328a4fd7be02743c975910d4 : SUCCESS in 56m 42s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca3b5a76507be8773e8a62ee1ffdaf507d686962","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-29 09:57:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d07795588ceee8394fbadce6ae25e9a9ac8e5ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 09:57:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e760718a8cfebc505c239cc6503c8dd0d249b16d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-29 09:57:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d9c37557a59174f7f43d14d966ba2066e322e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1132dcd86b4a2c944be4d0ecd276a2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52db674679f347e59f150f412cfff905 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352eaea18a2d46edaad5884ef0406624 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd5ec7880bfe474aac9d90f9bec44a6d : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/169639b72c0049c8b7606ca1ad0c9db6 : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34351feb7a0e4c2d8fb49dda33e6cf51 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6fc1f2fe524506902a5099ae7c069a : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/012e83d381104bdfa4c840c473b72816 : SUCCESS in 6m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/05250aaeb28e4391b9030167aba867f8 : SUCCESS in 11m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d286534c46d34c17b563b414721214bd : SUCCESS in 8m 24s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/57af5fd765334a72be4af06487d5f6ec : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16bb9de5c278cc970a2e8f90d9016300342637a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:16:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daf181bef3434754a231e85f1ded4f93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8132df7bad7c432da8914640cfd909a0 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2b2d90b3df743c4b5b491c435941c63 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6afc78f02eb04ea6b829e0a3eac09793 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aae2ab5894e414d8f9eaf6c0036fe65 : SUCCESS in 14m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/864609639420418696e4570fb843a8e3 : SUCCESS in 7m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/14505c702ca84d6589d02cc47ab78071 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"42a28048806c59ba6fce936991a06b6e1445d0b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c9e106f3d487c89b3c8e39d9ba5bbbd00b7d948","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:18:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccb85d47f884a93b3143c9abc668a79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83968e9d4b7843a9a9716bd49ffa51ba : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5ae97e14bc04a29a6fd68ede39d39e7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"e2e2f567e213105721e90b2f03908fb298acb3c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b61435b7e8a3485fbe96266abfb9da09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82c7ff7a06c4491b96c430280531a0f6 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3881ce6f885f4c9eb594f26df1451955 : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0745d4f65806b5c48f948ef40c4b37362143483","revisions":{"e0745d4f65806b5c48f948ef40c4b37362143483":{"kind":"REWORK","_number":1,"created":"2022-08-26 12:45:33.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/854549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/49/854549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/854549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/854549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/854549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/49/854549/1"}}},"commit":{"parents":[{"commit":"9a81eb268d27d7a74cac9ae15e40033dcd3a4f33","subject":"Update .gitreview for bugfix/21.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a81eb268d27d7a74cac9ae15e40033dcd3a4f33"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-26 08:42:15.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2022-08-26 12:45:33.000000000","tz":0},"subject":"redfish: fixes usage of ValueDisplayName","message":"redfish: fixes usage of ValueDisplayName\n\nIt\u0027s spelled this way, not DisplayValueName.\n\nChange-Id: I170d78bdb7ed0f6c36a80a9f2ceb9629f44394ed\n(cherry picked from commit 9f1f58c6af3aabcb3ae56e5f7b292f07e81e2864)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e0745d4f65806b5c48f948ef40c4b37362143483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e0745d4f65806b5c48f948ef40c4b37362143483"}]},"branch":"refs/heads/bugfix/21.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
