)]}'
{"id":"openstack%2Fironic~855230","triplet_id":"openstack%2Fironic~bugfix%2F21.0~Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","project":"openstack/ironic","branch":"bugfix/21.0","topic":"jsonschema-bugfix/21.0","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-31 00:24:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","status":"MERGED","created":"2022-08-30 15:31:24.000000000","updated":"2022-08-31 00:26:12.000000000","submitted":"2022-08-31 00:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855230-jsonschema-bugfix/21.0","cherry_pick_of_change":855166,"cherry_pick_of_patch_set":2,"meta_rev_id":"327daabdb635f48da124bc9be98c86cafb4f9612","_number":855230,"virtual_id_number":855230,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 00:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-30 22:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-30 22:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 15:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 15:55:20.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-30 17:55:36.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-30 22:52:45.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"ebdca1915a8ef073e1b3549afea7dfa56731c15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-30 15:31:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b6f01517014c81f09edc0e96f6d0ec9ad0e77add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5f6970455349d49554b3df62cba78d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd7e9764805c4bf4af96d896dd918043 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a926b616424c57ba54e3589733da93 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c1ab6ea25940f985b0260287d5fa70 : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b50f0e0b704a069fee660ea3e34167 : SUCCESS in 10m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ba18a70e7484b0cbfd0fc09a5fe5bc9 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f23fd1c022e4cc6ae50165b81af7a70 : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a6ae9bff7ef4ffbb6f066b89ad84bd7 : SUCCESS in 7m 15s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/8bcdbd85431f422eb83ce66413571b7e : SUCCESS in 10m 31s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a06807ee1cea4018869e36a18516da8e : SUCCESS in 8m 15s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/964b2e43ea4d468298ecdf2ffa6d9547 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"629b854dd6104c1376281763a0be72567e989a14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818030b7b20b47deb1834b0f7bf6d74a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c28bb50553264d13a4d60dd10f9a34b9 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15178c1d99a54a02b5ec628ae2d2bac1 : SUCCESS in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"307fd2ddc38c9ac7e8fa34d68fd4df7cc6f75075","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-30 15:55:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/a1c1417/a1c1417.html : FAILURE in 6m 59s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/5305c04/5305c04.html : FAILURE in 6m 18s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/3f401f2/3f401f2.html : FAILURE in 5m 14s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/d33dded/d33dded.html : FAILURE in 6m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/29d63d9/29d63d9.html : FAILURE in 19m 44s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/30/855230/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/378f8b2/378f8b2.html : FAILURE in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94f61cf132c9dc38fbdbca703c689804f8b3350e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-30 17:55:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/20f6aa4f5875453bbc5a67c8401e3e63 : SUCCESS in 1h 00m 12s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/003e00d5c12645a796589d8a4320cc12 : SUCCESS in 1h 02m 30s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4e3d1ce950444c1b8bf08147f49a44e8 : SUCCESS in 56m 53s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/05cbfecdbc6b42629aafd3e9995b3eac : SUCCESS in 57m 56s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e8cc5972190a43828a26a761647ff1eb : SUCCESS in 57m 11s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/be1204a367cc4f5b996e429ddad39760 : SUCCESS in 57m 10s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9b574abf9bea4fb9818ed1fe58715fc9 : SUCCESS in 56m 09s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/46e2a74054554287bb3d37c9fed7180b : FAILURE in 17m 15s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e748db55381e4c249d722fa4fbb35003 : FAILURE in 21m 53s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8dc6cb08ac514eae9af768cf90dbebdc : SUCCESS in 51m 58s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1ae87182f6ae464fbeb878ce6a96434d : SUCCESS in 59m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"bdaf9896b70462e069416ba74e4a24a13687cd04","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-30 22:52:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d970abea6058ed35f91e2da56932109c858af816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 22:54:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb0aa3ef263f8d908ccd9d3fd297d1b70bdbb5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:24:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f1bea321fab4dac9196040c3ec71618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f355cf2bfd42f598ba4f36f4522751 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab6659184674a3781d8f932277685b7 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9adc4e807ea9471bb6fca6f77bf9c776 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6fc630c98a40cb8a2e68849ceba912 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dbabc8b45894a4ea3396b0bbda0c5ca : SUCCESS in 8m 20s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4b9f32746d234abebe909d0b2b35a8b8 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6992cf8a44bc915d4fe07a8bb8f3d74a184c6240","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:24:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"327daabdb635f48da124bc9be98c86cafb4f9612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1af3ba12c48940adb30a28e42cb327c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd1ae74b9a24e20a1971105af5d4e0f : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67ca60012c34bc5a38ab6f3070a3589 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46fea705ad7f5601de51aa6840e38d043c3e8eb4","revisions":{"46fea705ad7f5601de51aa6840e38d043c3e8eb4":{"kind":"REWORK","_number":1,"created":"2022-08-30 15:31:24.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/30/855230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/30/855230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/855230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/855230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/30/855230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/30/855230/1"}}},"commit":{"parents":[{"commit":"e0745d4f65806b5c48f948ef40c4b37362143483","subject":"redfish: fixes usage of ValueDisplayName","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e0745d4f65806b5c48f948ef40c4b37362143483"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 10:15:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 15:31:24.000000000","tz":0},"subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","message":"Improve error message heuristics with jsonschema\u003e\u003d4.8\n\nBefore version 4.8, jsonschema did some wild guessing when producing\nerror messages for schemas with several equivalent subschemas. In\nversion 4.8 it is no longer done, causing error messages that are more\ncorrect but also more generic.\n\nThis change restores guessing the potential root cause without claiming\nthat it\u0027s the only possible root cause. Also the traits schema is\nsimplified to make it less ambiguous.\n\nSee https://github.com/python-jsonschema/jsonschema/issues/991 for details.\n\nChange-Id: Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4\n(cherry picked from commit 62f9c61ae6657227fd1f5dbc86c59a22b6bac28f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/46fea705ad7f5601de51aa6840e38d043c3e8eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/46fea705ad7f5601de51aa6840e38d043c3e8eb4"}]},"branch":"refs/heads/bugfix/21.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
