)]}'
{"id":"openstack%2Fironic~855231","triplet_id":"openstack%2Fironic~bugfix%2F20.2~Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","project":"openstack/ironic","branch":"bugfix/20.2","topic":"jsonschema-bugfix/20.2","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-31 10:48:09.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_10239\u003e by clicking the attention icon","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":[],"change_id":"Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","status":"MERGED","created":"2022-08-30 15:31:35.000000000","updated":"2022-08-31 11:36:15.000000000","submitted":"2022-08-31 11:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855231-jsonschema-bugfix/20.2","cherry_pick_of_change":855166,"cherry_pick_of_patch_set":2,"meta_rev_id":"bf847f5f2d74701115919d2016db1861b9acfb38","_number":855231,"virtual_id_number":855231,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 11:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 10:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-31 10:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 15:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 17:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 18:03:03.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-30 19:16:55.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-30 22:53:44.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"CC"},{"updated":"2022-08-31 10:28:14.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"961811336e6e2c7dd27684495cb3c01ab7e3b692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-30 15:31:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"471c0fef7f1a56a835aa36b610cf5e8b60ce6155","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:51:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1185641d98f743fe97a595ffff2845b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd02fb5e92734ebfabac96c279e2561a : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40c964d4d39841b8bcfc82c221cd3298 : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e623117187a3763c90a0d26d15e4acd621c54984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:22:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fce85fc76844d23bc6d0e857119a416\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c14293850d704584a988b8ec3d687398 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d101e6aa234d91ae48427826eab1ad : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de80a1ec5542458493f5b2a994797778 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1922778decd47278560c3ee6d64362b : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb689ba446d443588a2cfbb574a7b523 : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576dded2b5814b75b7ab9588f439608d : FAILURE in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fcae03ff3bb48e8a6352921f833a93f : SUCCESS in 7m 45s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/3991e36a5ade473fbf76121b8204f9d8 : SUCCESS in 10m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3971bc7a76e245cf94dc2618eadef8a6 : SUCCESS in 8m 50s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/93ba113c5b06474e8d21cee0fb2eae6d : SUCCESS in 35m 36s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/04767217b8564bd6a024f27ba18dea25 : SUCCESS in 29m 01s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/493b0373ed50476eabe28294c16b4a7e : SUCCESS in 1h 07m 40s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f119575b12d847e79fcb500054059328 : SUCCESS in 51m 27s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d75c32f501fc4d15b49f718aeb5b034e : SUCCESS in 1h 00m 53s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/40cc0989c20644e9b7422ee0baacb0f8 : SUCCESS in 1h 04m 19s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e1e5708f184e47c7adbc4c34bb97f4f8 : SUCCESS in 54m 32s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d6b908325f154617804e4a62c97af4e2 : SUCCESS in 1h 39m 47s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1f4163eb7bfb442a8b2c3dd74e66561c : SUCCESS in 51m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1c96f1e0c0814bd5b23ca3f608cd4c61 : SUCCESS in 1h 45m 41s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/2a94c0add28646bbbf0a1ea31f910a12 : SUCCESS in 1h 11m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/386e9d93ece8450d80e2f2f63962c5cd : SUCCESS in 40m 08s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/01f491ce7ce64e2e8960b2072c0a6164 : SUCCESS in 1h 00m 59s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/97b99b8b9a404a478924184d049b2f87 : SUCCESS in 7m 12s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7703469d84c34a599cfa053fc0584f55 : SUCCESS in 1h 10m 38s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a6ed2b36cf8a4e9183d95ab8edbf31ba : SUCCESS in 56m 49s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/bca4c34a027b4e7eb9a70374434ff77e : SUCCESS in 58m 43s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/2031ab3e56dc4f73843fd856cb0e6f79 : SUCCESS in 1h 08m 19s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/ecf106c837494bdabd09c57a706f2bbc : SUCCESS in 1h 13m 24s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b22e36a8d50d4663a16b9acb7427e941 : FAILURE in 1h 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d73fa6e9ce0d73404922802c9a2ce3659c99464b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-30 18:03:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/8203e07/8203e07.html : FAILURE in 2h 18m 21s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/45145e1/45145e1.html : FAILURE in 4m 48s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/4371349/4371349.html : FAILURE in 5m 21s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/a9b0457/a9b0457.html : FAILURE in 4m 48s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/2f70c9f/2f70c9f.html : FAILURE in 4m 54s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/855231/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/8815874/8815874.html : FAILURE in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af8b404c411a11f12f263494ecb8193fdffee0d7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-30 19:16:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2494128eb2474add876f771914ca30d5 : FAILURE in 29m 02s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9d837d968ecd49569ea200cfb07d2d8d : FAILURE in 26m 56s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7fb7fc2660b8446f81e4901627237c36 : SUCCESS in 54m 34s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b2764163b8434b888dc2d71486daea4d : SUCCESS in 52m 08s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e8736c1f321b40a189b40e82f1b3c54b : FAILURE in 25m 22s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fbcfaa3441db4361b67a1a1187115afe : SUCCESS in 49m 56s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9b36d5be30ca4b53b3e716c589372796 : SUCCESS in 49m 42s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a73617d1ebc94216b15739434c3b6b10 : FAILURE in 24m 55s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/06915a80b4b34410b65e4ddf26d10a76 : SUCCESS in 55m 48s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/677c98c19a964ff7b207075cbb04a3fb : SUCCESS in 48m 47s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/89c1c6c6f6924e7e954e19ee2fa2bdb2 : SUCCESS in 1h 22m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"d979cb5dc51e5a8a2b28cab67e5ffb715dd24028","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-30 22:53:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d480dcb92f4b146b6c88e4f6999ee150bfacb28e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 23:16:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6403a3e31ad48f68823cd943bdf266b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b624aeb934bd4979a7fee044f8247d2a : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8170ba35d2254355b1999134d94a3581 : SUCCESS in 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c459e4d9282016d0c68b3496988b046afa27c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:33:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ea3bdb22aba497ab263da6c9022c761\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bfc9528deac4c6e84025a60144aedd9 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19f0dabfe7b4e0bb7d937e35e341f1f : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6857739f4d9f4b4c9d2ab058b5b1b7b9 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11ca33f9627448f1821896aa720e3d20 : SUCCESS in 11m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e9e4ff94a3b4642933ffafb6a8b60f9 : RETRY_LIMIT in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66e88799aae42cd9e95657cefb91426 : SUCCESS in 14m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597ab7534ffb4526acb4e6f5df4c760c : SUCCESS in 6m 13s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ff3eeffe85b04459b3766dc65b71bd1c : SUCCESS in 10m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4d6f037e6d984435aa1a6b4d44914998 : SUCCESS in 13m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/bf531dde6dbf443fb930d5a017bb3022 : SUCCESS in 36m 22s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e523d019e01f4a00b1dc2919f91b2540 : SUCCESS in 36m 24s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1d2241034ee84f428ee187810eecde9a : SUCCESS in 43m 57s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/7c34e63352814c1688b0d2b10edef600 : SUCCESS in 1h 04m 28s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/222a42906e2144e09158e19857372ef0 : SUCCESS in 43m 26s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7567840d67a74110928215a92aec4153 : SUCCESS in 1h 00m 45s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/12e7522ec32846a69d4ae7e1e4602f11 : SUCCESS in 36m 08s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/0b15dace0e914fa9a1d73aa14043bbfa : SUCCESS in 1h 32m 18s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/1e9e85c4d1344c1c942efdbcbcc228ec : SUCCESS in 58m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/38932ee7a6784678a067abc670ee72f1 : FAILURE in 1h 29m 58s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ff7c0e8fc35f43f0b0abb41c3203981f : SUCCESS in 1h 09m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/13428d9bc71a4c1d857ac7bc78d16790 : SUCCESS in 44m 11s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/25b322b9de2d49f19e4ffd7588141ec8 : SUCCESS in 59m 35s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/56ab1d6dc47c4f9abacf4caae8b55f1d : SUCCESS in 4m 43s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/91d6831709f24ae199332893acad4b7a : SUCCESS in 49m 30s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7dcf869016bf4748bdd2aae8ea4762dd : FAILURE in 1h 09m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/7b5944560f104d1c91914d43a7726a38 : SUCCESS in 48m 07s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/7d31903c4eee4fcca31e035326dcede1 : SUCCESS in 1h 13m 10s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/df095859b00644b7978dc00291ffc22a : SUCCESS in 47m 14s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/a50e1d2f38254e019113ed9ca5f46866 : FAILURE in 1h 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a172dae48afc7467d27df277bfe98d3594df47a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-31 10:28:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfb50a3932305c78a4a2c64f02540e31fe6abc15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:28:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1da6491749a094d80f7eecac656aa3567aed87d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:34:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d8aadc401904920975dadafc6cc8293\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862a7413fbce4f45abd5f3f8b83991ce : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/061755e42acf40b9bce7c518c01e9bcb : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb8737e7d05e4275bbf6c8eae479635e : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4a5abe6b8d49a38329930823a3a11e : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d7a3721f1047d199c909cb07283eb8 : SUCCESS in 6m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/dbb44b2a04784e1dba98857d1805ecb0 : SUCCESS in 11m 26s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/37a12b914b1d4ed7862eec4d16b8dbd8 : SUCCESS in 30m 02s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/ef66da91de3646279e02a11b9b6339b2 : SUCCESS in 27m 18s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/fe9e08213e6844219edf36536287d29d : SUCCESS in 51m 03s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/744985fe9b9249218cf0ed0e931cf512 : SUCCESS in 1h 00m 29s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/499594a6aa2f4e15a72989c9c36a67e0 : SUCCESS in 47m 49s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2c58fdf5dc7d44feadf3c339ec5a001e : SUCCESS in 55m 21s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/5b061f192b5c445fb0b68d490d163f96 : SUCCESS in 42m 49s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/6c3d1b8c71b34c0e9fe24208990f0491 : SUCCESS in 1h 05m 05s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/07ff76a3f0ba40609e556c45d82d27fc : SUCCESS in 50m 42s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/94509fa50ac140eeade58558b605ffd5 : SUCCESS in 59m 18s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/81590f0963c7441da48a39c7d179aa59 : SUCCESS in 43m 22s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d037587c67f0457d93b14691f2433fbf : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"80f5bf9c67c6d1ca61249a8d7da1f3de1f066689","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:34:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf847f5f2d74701115919d2016db1861b9acfb38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:36:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3510aa4a75244cd3ae3b066fad03d7e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db070a519c5b4dc0bcb8155a965cd5c0 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e7b67ff622d4e748e2e29e2b8e176a2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e302b79c798901ba87542e674e6fae08ced7abe","revisions":{"0e302b79c798901ba87542e674e6fae08ced7abe":{"kind":"REWORK","_number":1,"created":"2022-08-30 15:31:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/31/855231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/31/855231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/31/855231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/31/855231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/31/855231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/31/855231/1"}}},"commit":{"parents":[{"commit":"3349d7e6c92c324e69d7ebeeba5bf24fdf42d707","subject":"redfish: fixes usage of ValueDisplayName","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3349d7e6c92c324e69d7ebeeba5bf24fdf42d707"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 10:15:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 15:31:35.000000000","tz":0},"subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","message":"Improve error message heuristics with jsonschema\u003e\u003d4.8\n\nBefore version 4.8, jsonschema did some wild guessing when producing\nerror messages for schemas with several equivalent subschemas. In\nversion 4.8 it is no longer done, causing error messages that are more\ncorrect but also more generic.\n\nThis change restores guessing the potential root cause without claiming\nthat it\u0027s the only possible root cause. Also the traits schema is\nsimplified to make it less ambiguous.\n\nSee https://github.com/python-jsonschema/jsonschema/issues/991 for details.\n\nChange-Id: Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4\n(cherry picked from commit 62f9c61ae6657227fd1f5dbc86c59a22b6bac28f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0e302b79c798901ba87542e674e6fae08ced7abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0e302b79c798901ba87542e674e6fae08ced7abe"}]},"branch":"refs/heads/bugfix/20.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
