)]}'
{"id":"openstack%2Fironic~855232","triplet_id":"openstack%2Fironic~stable%2Fyoga~Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","project":"openstack/ironic","branch":"stable/yoga","topic":"jsonschema-stable/yoga","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-31 00:24:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4","subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","status":"MERGED","created":"2022-08-30 15:31:50.000000000","updated":"2022-08-31 00:26:11.000000000","submitted":"2022-08-31 00:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855232-jsonschema-stable/yoga","cherry_pick_of_change":855166,"cherry_pick_of_patch_set":2,"meta_rev_id":"bac8616ad1773ad8b930b15fd705e4ceb098a95c","_number":855232,"virtual_id_number":855232,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 00:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-30 22:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-30 22:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 15:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 16:21:48.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-30 17:19:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 19:30:34.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-30 22:52:38.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"cac3b2cc2b3c2b85ece3b4562a5128ad247cf811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-30 15:31:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"af4a64f7edb7db7b4e239d27da0be1161b3c9691","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4e547744dfb42fe86dde9d1031941da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8f8cf294c1d4c1abb9750cc2a3294de : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2805e2f0cd34d379af686ca85a540af : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90f09a856fa28dd08854db5727ddc3a58ee59e39","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-30 16:21:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/9288c20/9288c20.html : FAILURE in 5m 14s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/5a3f5f0/5a3f5f0.html : FAILURE in 4m 50s\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-ilo-fw-update-direct-python3/cc785d8/cc785d8.html : FAILURE in 4m 49s (non-voting)\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/de34a53/de34a53.html : FAILURE in 22m 49s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/d634286/d634286.html : FAILURE in 15m 53s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/855232/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/8515d3b/8515d3b.html : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b098dc69ed9e0137fc54d3a5fea99122b45bca4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9978297336cd46c497b107e8362084a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5517ecb9166442ca80ca3474034e48ad : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012161478ae240f897ed231e738fc33b : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/208a8849d5da4a049658526bb640a6a7 : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fba2df2602849b7adf526b63bb014c6 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140653db680b49db821035ecf580022b : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f8b9d44728c4c71ace2b3f239c437cd : SUCCESS in 8m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a1e679651bf3432299c03f1017920517 : SUCCESS in 9m 27s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/20b69ecceb7246c08219d8fe4531e179 : SUCCESS in 10m 38s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/19b881d41b2a4fe48c4ae0ecba390bd9 : SUCCESS in 24m 12s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/6a49aaa7ddb7426cae2fbea591d1f10b : SUCCESS in 28m 48s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/284be037f72a4d0bb4610964f7173748 : SUCCESS in 1h 05m 15s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1b11572941d14f73af7c7440bfdbd480 : SUCCESS in 56m 48s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/544fa34151b24a0f94658b0e283cc716 : SUCCESS in 44m 54s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f34be6cc3a40450b98af34cde845230f : SUCCESS in 51m 07s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b850e0272dce4b779edd2c0c13f43ec4 : SUCCESS in 50m 02s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ed340e8cf16a42eab43b3d583970ae66 : SUCCESS in 43m 41s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/6d1308eb67bc4dc0a15be61942211b5d : SUCCESS in 1h 24m 46s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d43ac463ffa9458f9a43990f551a8f37 : SUCCESS in 51m 44s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1ee457ef5cd54fb6a5123599bb46347b : SUCCESS in 1h 40m 13s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/5dc3f0107726443493d1554346282f5a : SUCCESS in 1h 00m 45s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/03cf81628977433a8686202773acc96e : SUCCESS in 41m 58s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/b788a7733dce414eb76cf7edde693212 : SUCCESS in 46m 35s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/b46d82e5465041fd91078b5609cc757c : SUCCESS in 1h 34m 31s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/2b8cc401e4dd449dad9ca997945f64f7 : SUCCESS in 5m 42s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3e81d1083b6e43fab0bc4e3987d719e0 : SUCCESS in 54m 54s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5c88afaf649e4b0795bac18bf3a71050 : FAILURE in 58m 19s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/694c48baa94545708bec6977da7496d9 : SUCCESS in 56m 14s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/97630016930c4352b29b8e3804c5f000 : SUCCESS in 44m 31s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/106c1b1a293b4af1831e64d91f289875 : SUCCESS in 56m 42s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c028db7cc04567a8f6d7756967dd55 : SUCCESS in 25m 45s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/bab73cf23fb6472ea7b02e1b60070d83 : SUCCESS in 30m 01s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/61151c6ac83146b88a3dc11dd403bd07 : FAILURE in 1h 22m 44s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/7785eb5fac1248a6a995f1f15c766b0a : SUCCESS in 27m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"288114c0e1dafdbc23dcf14c09c4876954ee2849","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-30 19:30:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/182a492f472142de867400f8f72f348f : SUCCESS in 55m 43s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d2e23e5ccac849278ff8cd00aeae9030 : FAILURE in 14m 16s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c8800fab7c4d46548a785c39ebc7a45c : FAILURE in 26m 35s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a0c499b8d43f4c81b3885b9919adb119 : FAILURE in 27m 08s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/de6d2be84b084ed7b7815cd4863bade6 : FAILURE in 27m 10s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ef0bd9eeedc049828ad72d008d4968c4 : SUCCESS in 1h 00m 49s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0915d33335894ab194494e754e9d201e : FAILURE in 21m 22s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/53411390284b406c9d4cf2edc8beafd6 : SUCCESS in 50m 03s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/40e465def73e4845b1a1c1854b54aa73 : FAILURE in 26m 36s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b6af8052aede496c859f32c2ffc4c5de : SUCCESS in 51m 47s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3762fdc4fa4445dcae2d5bcf85265242 : FAILURE in 29m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"648edfb97a35153226e852d807d52b1a68283b1c","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-30 22:52:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2eb3c61506944b1835c6f875c99d7c783fc69d4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 22:54:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1125a32ad07573418e29c570d4e48782809b7e4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:24:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc797c37e4c340fab645a9a3c408ff51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c33840316d44348f295e5161bdbffc : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cab90c16fae84086a793c915d2165ca4 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75971d7d4cc240cbb736d02a8f8f3b0b : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb9e90ddf664289a64153aa672524a5 : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b892b245da3415085aa09a300968efd : SUCCESS in 6m 27s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/7a79154f24534cb58ffc3dd43a268eee : SUCCESS in 10m 18s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/69b3e06efe91493fb83d906bbc43a30c : SUCCESS in 23m 17s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0e0abdd207864556b387919434d07439 : SUCCESS in 24m 36s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ff3773d69aeb4f9e80740a9506619c72 : SUCCESS in 1h 11m 20s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/5da20567b2264731b5bb499c751ea514 : SUCCESS in 45m 19s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/2d950c2648b74c2d9dde438b41ed5e6d : SUCCESS in 49m 39s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d970ccf6e81e4ec5b9278d321f7e6fe5 : SUCCESS in 46m 58s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/efbf32294b404376a2f02f5772e1c534 : SUCCESS in 51m 03s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/dda17e7b6eb04845b49f064240ced5bf : SUCCESS in 39m 05s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/5b23f94b06d5405c9efd0153da4da1d5 : SUCCESS in 1h 07m 56s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/fdb9da1dbcab46729f7fec012d016ab7 : SUCCESS in 50m 50s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/819e7146ae534ea68171d94e002a610a : SUCCESS in 52m 36s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e57d0df9d5a44687861f9a0c14ebbcaf : SUCCESS in 41m 46s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/cea2f795f225483994531c0573a8eeca : SUCCESS in 45m 22s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/dd4054686a494ee49525262d82acd59a : SUCCESS in 1h 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ec5bb5bbbcc8aab07ba8a57e8e569390aac5da18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bac8616ad1773ad8b930b15fd705e4ceb098a95c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:26:11.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/883d37d481374611be65d9e9cbdf48c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f5318ac93b42a19ca988e9b1a60633 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efad80623f634bbab50f2abcb1f0224d : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41b01ca44e8e98d698c2b568757806a57aabbb1e","revisions":{"41b01ca44e8e98d698c2b568757806a57aabbb1e":{"kind":"REWORK","_number":1,"created":"2022-08-30 15:31:50.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/32/855232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/32/855232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/32/855232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/32/855232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/32/855232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/32/855232/1"}}},"commit":{"parents":[{"commit":"4bcdd4d90380aaa744f6a1c9b16f0d495892ab61","subject":"redfish: fixes usage of ValueDisplayName","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4bcdd4d90380aaa744f6a1c9b16f0d495892ab61"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 10:15:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-30 15:31:50.000000000","tz":0},"subject":"Improve error message heuristics with jsonschema\u003e\u003d4.8","message":"Improve error message heuristics with jsonschema\u003e\u003d4.8\n\nBefore version 4.8, jsonschema did some wild guessing when producing\nerror messages for schemas with several equivalent subschemas. In\nversion 4.8 it is no longer done, causing error messages that are more\ncorrect but also more generic.\n\nThis change restores guessing the potential root cause without claiming\nthat it\u0027s the only possible root cause. Also the traits schema is\nsimplified to make it less ambiguous.\n\nSee https://github.com/python-jsonschema/jsonschema/issues/991 for details.\n\nChange-Id: Ia75cecd2bfbc602b8b2b85bdda20fdc04c5eadf4\n(cherry picked from commit 62f9c61ae6657227fd1f5dbc86c59a22b6bac28f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/41b01ca44e8e98d698c2b568757806a57aabbb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/41b01ca44e8e98d698c2b568757806a57aabbb1e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
