)]}'
{"id":"openstack%2Fironic~856004","triplet_id":"openstack%2Fironic~master~I95178f240d3757b869605c7f83ca30785435c305","project":"openstack/ironic","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I95178f240d3757b869605c7f83ca30785435c305","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-09-06 04:05:46.000000000","updated":"2022-09-07 08:37:24.000000000","submitted":"2022-09-07 08:35:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856004-zanata/translations","meta_rev_id":"a544157a4bea7ab7e159b2e108236ed22b99f7ff","_number":856004,"virtual_id_number":856004,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-07 08:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-07 08:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-07 08:22:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 04:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 08:22:52.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"f8aa5d5c0ca94426b3785947c471808b5b878fc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-06 04:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c1e31a9b2b6180e3678f04c4bc54008de6e2b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 04:25:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f6c1f77ce14381bc17ea485325def4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30273d9027314bf1809be8231c5da46e : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e42dc8a0b864cbda006878aac90cd85 : SUCCESS in 14m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2960366503a240e5a06f86cd2f59a571 : SUCCESS in 7m 50s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1f99bf2da3f946db992ece1c587b0c63 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9958add83f0006c73bbe6c3a1e82d31e745dee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-07 04:21:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1af4154c7b82768a292021e1340b244b6366f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 04:34:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16eb98e45ad4067b7aa72570970cccb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f3d1cfaca646c79a8c27925bba9a03 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462188a5ebda4e5aafd4272572c9e97f : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43b2c4eeaf9d4f6aa3ec15775ef1251f : SUCCESS in 6m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f2c7ef2159724f6ebb42952630c88d87 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5b43913650623cede055f3617430bac359a0d99b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-09-07 08:22:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6208809e17a62e06870322ace7ad593f573a76a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94dfa91d001b8b6f5366c7b094deb14aeb7918fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:35:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72bcd98e241e4d1e88ef2fedf985f6b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c5d3bf766f848fea0312af6b4045d64 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc2d4e8cd4948408b86d3d018631c08 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e98e9077977403180931e8d5891501a : SUCCESS in 6m 28s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d2c334dc47a146139e3d1daa4b7256e0 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"76db402145de00ac394bb5a86542b75a863debfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:35:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a544157a4bea7ab7e159b2e108236ed22b99f7ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 08:37:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57304f95028c480b9fb8a8bcb1aa684b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3db637707444ca6803b027337fa388d : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/527649a4b396453b8a0f470827c686c3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b796d7b833f5e8589401418e0c4d967bab7a7d81","revisions":{"b532747a16b5daa732b4d613e525e606f67a0547":{"kind":"REWORK","_number":1,"created":"2022-09-06 04:05:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/856004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/856004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/856004/1"}}},"commit":{"parents":[{"commit":"d5df494ad51ce942572b13c415cdfe645f2a7b74","subject":"Merge \"CI: anaconda: permit tls certificate validation bypass\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d5df494ad51ce942572b13c415cdfe645f2a7b74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-06 04:05:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-06 04:05:41.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I95178f240d3757b869605c7f83ca30785435c305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b532747a16b5daa732b4d613e525e606f67a0547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b532747a16b5daa732b4d613e525e606f67a0547"}]},"branch":"refs/heads/master"},"b796d7b833f5e8589401418e0c4d967bab7a7d81":{"kind":"REWORK","_number":2,"created":"2022-09-07 04:21:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/856004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/856004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/856004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/856004/2"}}},"commit":{"parents":[{"commit":"9a01abbd71a4e1c3b2d42eb78e51ac21387acbc9","subject":"Merge \"Fix compatibility with oslo.db 12.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9a01abbd71a4e1c3b2d42eb78e51ac21387acbc9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-07 04:21:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-07 04:21:13.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I95178f240d3757b869605c7f83ca30785435c305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b796d7b833f5e8589401418e0c4d967bab7a7d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b796d7b833f5e8589401418e0c4d967bab7a7d81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
