)]}'
{"id":"openstack%2Fironic~856345","triplet_id":"openstack%2Fironic~master~I7c26c20e4b36c4f3b98873939677b966ec6186a5","project":"openstack/ironic","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-20 16:11:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c26c20e4b36c4f3b98873939677b966ec6186a5","subject":"tests: Add a WarningsFixture","status":"MERGED","created":"2022-09-07 23:09:34.000000000","updated":"2022-09-20 16:12:59.000000000","submitted":"2022-09-20 16:11:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856345-sqlalchemy-20","meta_rev_id":"744c492cfbf70105f894e631003f448493e7b0b0","_number":856345,"virtual_id_number":856345,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 16:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2022-09-20 15:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-08 11:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2022-09-20 15:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 23:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-07 23:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-08 11:47:52.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-09-20 15:56:08.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"d8749672567ad14ad2656564e3e0017543ac9f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 23:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26085896385d89d33c09df00abbd08cf03cbf796","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:28:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8acb9bb37a0243aa92bbe9e6e1e5cc26\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06ce8d1d21a14eac9d9593dbddce60c1 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dedec8adeeb948f9b4cf8e84f4d4c60d : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf1994d5589cc656eebfb5f1117cd3f693bf62cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:39:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f9f5791fb443d6a7cceb03e9c8c952\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/838d4b10e6434757b3402ae4ecce1582 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32cd30aaea524bdbbd70d6aa0b815a7e : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb79ffbd5db485c9e592400bf8d3842 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/532ec9843e1d4034bea4985c438f7ad8 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd2ef90413b4f6abc7e24242585a69d : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9052877496e4c28970da05948aea87e : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627c613e680c4757a8aa6491eee26bfa : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0b84c4c8c86422883bd442b63c20673 : SUCCESS in 7m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/525ba3f10c2e411dbc000c574e4c2f12 : SUCCESS in 11m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2a94c2bb7b8a40c09b4e85e17470a4eb : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6e497692f70fa1d86b3ea05ead9bc662387d5ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 10:00:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d42f2de7dcdd953edb705df9ac90528588fdfd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:16:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90ab41a83b7340b8a6bbc08eb9237a91\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6dfafcc8b8a84752b0eabc6508b2df59 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eade8a4ac6e4daeba3a7a4003178def : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f9a23f572c47279449a0a227496779 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a36efa1ed2e243bebe8e15fb70c9ff45 : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b428217ba7ac405b90fe6124c3e6c7a9 : SUCCESS in 9m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcb0b64ce66e4930984e2ce4d91e9cb0 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819cb394418d42e09772aa0d07bac51b : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e831f2e31f6b44cc8f6eaf1856e79f8d : SUCCESS in 9m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6a64381490f944e2b40601bb7ddb7fcf : SUCCESS in 10m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/94f7b94b6f27453e997e3eecc3389de3 : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"211712d8f2ba5d7a4cee8a08218061b3fd8b87e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:24:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8954917de096405a9b59bdd6952acad8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fa19a69a0e247b791b7796deb9ac7d7 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa42902964f6407a85bb33c4f345312d : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b13f89786714f7ede5876737f2e58dbb3741ca5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-09-08 11:47:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"153c1924ccf624237c1e3be9eff63d2db756df14","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-20 15:56:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c9c5dc786987bdd7831cf216bb21ceb3a3aa882e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 15:56:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aaf7f75c61c1381507d12b955c752030f7fcf425","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:11:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25aa792ef8e4592a68a6055cd2d8242\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9e60b33431954264a972d4f545b847a6 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1599bdded17410eade1753a52fb0d82 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c698920da264d9c8f37c59412d1355d : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5354fbc618fd4523b8eba0db27486f4d : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519cb41bdaea48e8bdc5f4ee6a15545a : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb70987a47b4414ea99f7bf691c8236b : SUCCESS in 6m 43s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ef43249698e5400492663060ab585c2d : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3e15306d5eda4d8fdd622a3bf43ad1efee4b9bce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:11:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"744c492cfbf70105f894e631003f448493e7b0b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:12:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53748dfd73c4e6a89482058fe218677\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3bf86983da594b898c6de4020bcc377a : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e94966e17b40cf863c52348ef38686 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d92cbdcd8f0a4292949137f5787a177b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25b3e6796dbdd5d53d2ac8aa34185c7cca138085","revisions":{"986af7e337d2ad61b62d106fe26145ed5e45c7c8":{"kind":"REWORK","_number":1,"created":"2022-09-07 23:09:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/856345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/45/856345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/45/856345/1"}}},"commit":{"parents":[{"commit":"b796d7b833f5e8589401418e0c4d967bab7a7d81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b796d7b833f5e8589401418e0c4d967bab7a7d81"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-06 15:13:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 23:05:30.000000000","tz":60},"subject":"db: Raise warnings on SQLAlchemy deprecation warnings","message":"db: Raise warnings on SQLAlchemy deprecation warnings\n\nChange-Id: I7c26c20e4b36c4f3b98873939677b966ec6186a5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/986af7e337d2ad61b62d106fe26145ed5e45c7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/986af7e337d2ad61b62d106fe26145ed5e45c7c8"}]},"branch":"refs/heads/master"},"25b3e6796dbdd5d53d2ac8aa34185c7cca138085":{"kind":"REWORK","_number":2,"created":"2022-09-08 10:00:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/856345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/45/856345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/45/856345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/45/856345/2"}}},"commit":{"parents":[{"commit":"b796d7b833f5e8589401418e0c4d967bab7a7d81","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b796d7b833f5e8589401418e0c4d967bab7a7d81"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-06 15:13:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-08 09:58:42.000000000","tz":60},"subject":"tests: Add a WarningsFixture","message":"tests: Add a WarningsFixture\n\nWe will use this in the future to prepare for SQLAlchemy 2.0. For now,\nwe\u0027re simply using it to filter out some of the more annoying warnings\nand to highlight general SQLAlchemy issues we need to address.\n\nChange-Id: I7c26c20e4b36c4f3b98873939677b966ec6186a5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/25b3e6796dbdd5d53d2ac8aa34185c7cca138085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/25b3e6796dbdd5d53d2ac8aa34185c7cca138085"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
