)]}'
{"id":"openstack%2Fironic~856346","triplet_id":"openstack%2Fironic~master~If8fc919d4a04325ca48a037c02481d9efe54279f","project":"openstack/ironic","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-04-17 09:34:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If8fc919d4a04325ca48a037c02481d9efe54279f","subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","status":"ABANDONED","created":"2022-09-07 23:09:34.000000000","updated":"2023-04-17 09:34:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"75fea7d3dcb75d82a4dc8bba483c006e082f0189","_number":856346,"virtual_id_number":856346,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-08 10:22:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":2,"date":"2022-09-20 21:34:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2022-09-27 09:35:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 23:30:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-07 23:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 21:34:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-09-27 09:35:33.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"e888b60d974720272d8f70b80a6ffbdb4589417a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 23:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7ea4a7d7425fb607f0a03eecd245ca51619ef45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:30:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423935830dbb4ea8ba0a5961b2b7ae80\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29e5462005bc48d0b77a9907fb15bb85 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be5577dd720b468c852cac2a775692f1 : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"782965aad759364aba74005b4425586bc1c42012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:41:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6375cf1f89cf4566b2523ea8d510fada\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d260bc2f9f2047f09f2633cdfa3a730a : SUCCESS in 5m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af21e6260784ff2aad332dc0459aa36 : FAILURE in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae69e07a3204c3fbb83385583980e5f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4853fe6a97944979a47afaf24c292633 : SUCCESS in 8m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d81b0cfbd214788bc8f4ff7249128ad : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c456672210854d12b138b773a5a2117b : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4887a3639dc42b3acae38818622b503 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f778992cf2284d739ec72692e1cec716 : SUCCESS in 6m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/711243c037a349048be65dc70910b491 : SUCCESS in 9m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/803243298f474095ab0d1d2b65c2aaa5 : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ec4945d0345843b3f41a13b68f53568e4019b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 10:00:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa878113d240d5e2ee3c2e1a80510441a835f252","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:19:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b47030fc0e439d8d0a91c781cb29d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6bb050e229947a1ba616512ec018a2c : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76ad9f2c77cd49a1a9fbaf1e53422d5d : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20a3f4cb97ce99be6c4a1e4e9104b95d99391145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:22:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d724836ae041908a9dafed4bacda6b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/268d405601de4aacb84a992a88c611bd : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a1c50b0bba4f2b8df44dbeb0b0c437 : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd96f2f8d8542f998e83843f0f159db : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/069f790079ad41fa8a666837099d9aef : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58db2de62386421ab4d7bdb960d6276d : SUCCESS in 8m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41fb783ea7f74be88bbfe3c3e22cc472 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8079f7f0b70d44d4ba9f3c15a6fc2abf : SUCCESS in 16m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6182b79a8bba4b6594943522728619b8 : SUCCESS in 6m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/9f9510921e334f1ca32a56f8a3b1735a : SUCCESS in 10m 34s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f9675b7e548545fd8c57925b016c7ab5 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4479907f3da168cba3f12ba306ddf7d068343fef","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-20 21:34:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44750d167d16f23bb7492481d9af331aa3bd6307","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-20 21:35:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2859d4ce24df688159935575975fed683c4aa35e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-21 14:28:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4c7b74240cc1f61d21237fcad215457f359b725","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-09-27 09:35:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"75fea7d3dcb75d82a4dc8bba483c006e082f0189","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:34:09.000000000","message":"Abandoned\n\nThis was done by Jilia in I651ec4b50c79be6aa8c798ee27957ed720a578d8","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d0a18d34c07bde81530460fee52626a897e7bb1","revisions":{"42edff6cd90f2c2a8253bbb20415e0b993d56e75":{"kind":"REWORK","_number":1,"created":"2022-09-07 23:09:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/856346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/46/856346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/46/856346/1"}}},"commit":{"parents":[{"commit":"986af7e337d2ad61b62d106fe26145ed5e45c7c8","subject":"db: Raise warnings on SQLAlchemy deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/986af7e337d2ad61b62d106fe26145ed5e45c7c8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 22:03:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 23:05:30.000000000","tz":60},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\n[1] should provide a guide for how to resolve the outstanding issues.\n\n[1] https://review.opendev.org/q/topic:\"sqlalchemy-20\"+project:openstack/oslo.db\n\nChange-Id: If8fc919d4a04325ca48a037c02481d9efe54279f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/42edff6cd90f2c2a8253bbb20415e0b993d56e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/42edff6cd90f2c2a8253bbb20415e0b993d56e75"}]},"branch":"refs/heads/master"},"4d0a18d34c07bde81530460fee52626a897e7bb1":{"kind":"REWORK","_number":2,"created":"2022-09-08 10:00:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/856346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/46/856346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/46/856346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/46/856346/2"}}},"commit":{"parents":[{"commit":"5825b29b10abe24ea169810b1d0d9b08f3204c62","subject":"db: Resolve SAWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5825b29b10abe24ea169810b1d0d9b08f3204c62"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 22:03:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-08 09:59:51.000000000","tz":60},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\n[1] should provide a guide for how to resolve the outstanding issues.\n\n[1] https://review.opendev.org/q/topic:\"sqlalchemy-20\"+project:openstack/oslo.db\n\nChange-Id: If8fc919d4a04325ca48a037c02481d9efe54279f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4d0a18d34c07bde81530460fee52626a897e7bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4d0a18d34c07bde81530460fee52626a897e7bb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
