)]}'
{"id":"openstack%2Fironic~856348","triplet_id":"openstack%2Fironic~master~I9aad5f3845f8abc64b1b2abc78b829929f73aafd","project":"openstack/ironic","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-04-17 14:52:43.000000000","reason":"Change was abandoned"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-04-17 12:41:53.000000000","reason":"\u003cGERRIT_ACCOUNT_11655\u003e replied on the change","reason_account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}},"hashtags":[],"change_id":"I9aad5f3845f8abc64b1b2abc78b829929f73aafd","subject":"tests: Comment out prints","status":"ABANDONED","created":"2022-09-07 23:09:34.000000000","updated":"2023-04-17 14:52:47.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"f1da0e38e5e5db5291e6e1e3526a71c39f148ff6","_number":856348,"virtual_id_number":856348,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-17 09:48:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":-1,"date":"2023-04-17 12:41:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 23:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-07 23:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 21:37:34.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"28e25eeaa095be13f75c3f5a7df9d9559dec4a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 23:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02708f697072e6f1f8b577c65dface17e9e9448d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed481e76ea794599b369b0032dfe4aa8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/414af5e3817c4302b62003275a5ee3a9 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f2ec4032d974c35861a76b6153f17e1 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b247bb10e8d07ad8c0ccf3adf53a89fa0ce1f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 23:39:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e42ec2cd27d4823baebf557379eb922\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c023e3fcef241c599e53de6cff4487e : FAILURE in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c7d8971f134af3a6bc6d1ac350a633 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b35f252abd248828b0cdd4bf7d4a17d : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bc4cf8d7b5440248ee4656e3c69ce26 : SUCCESS in 7m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2e1061d288a4456b7786ec575846cea : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ff6873278549359891875e8902ba38 : SUCCESS in 14m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e2543fbaf0a34214a709d216350556c9 : SUCCESS in 9m 18s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/5cb15965a3da4d88a2f0f70d6248a911 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f273072ce3dcec1b19539dcedc248924d6919c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 10:00:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4da97fc173f60f89d5b3b2606f818d37167ffdf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:17:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b8356711334285b328567210f4dc3b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5dda0761db54197b2f7edb10bc447fe : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227ef094eb0144e98c0b667cb8fd8bee : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cd10811d252438b96fec86f22345998 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/590176be589f4e5cad6116466908bb5d : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e60dc98876a94e99a931083222f5a9d4 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b686068a40f74faf94c3d9b6d01884fe : SUCCESS in 11m 30s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b53d18f57695429aaa94d14bc0a88c6f : SUCCESS in 11m 09s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/20f2a6b094cd46a0bbabc9d2574e2a68 : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb3a27ea57d7a999e603280682a56208cb42fcac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 10:40:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4aef710c33427fb09cee51e98a0030\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e56acf0648254c06be0175b81feffe85 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e783f02ce6414a2aaa306a726193a56a : SUCCESS in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c747a3ce4dfa7aadb5575b6beef0275284aae863","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-20 21:37:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3c9837f1f29115025f4ab22ee8e193adb61e6757","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-21 14:29:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb694634bda63e0de1120ac5d263ce5ab73294f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:34:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a9c6d9e2fcbcdf896f49a52b1decd7c38b3d1a46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 09:47:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc2878d876c742918174f07b0117652d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d576921e3530453bb3d6589fde68ad9a : FAILURE in 11m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1135820246d94abd97f095546d4186c8 : FAILURE in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5ba18a37927c9906bed28fb1e259dc086e49bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 09:48:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41340a63d6064d7cb3b6914ca86e0145\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16020f503c7e48ab9e4703765a18decd : FAILURE in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f949890f06354b6884b3dd4d1dfb3822 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/342dcfca134842f0bf38deaaac36fa2d : FAILURE in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89f898a96e9427fbf87138b4048f937 : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd2931502e54bdbb671c0775f4359d9 : SUCCESS in 11m 41s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ea432278ee094cbc8dd7448b54f6d51b : FAILURE in 10m 25s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/acd0c8794329419582c561a77bcca38a : FAILURE in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"46b93992eca18aeba532e728331f636a82cfe5ee","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-04-17 12:41:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5e8b7247946322bf75c722e5f5de0c2f0dcfc9d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 14:52:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"f1da0e38e5e5db5291e6e1e3526a71c39f148ff6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 14:52:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5a644361b025cb5bcdcb7288c40e5553500b4f33","revisions":{"e9ae06acb0179042632857f5f82d9f2adf98332d":{"kind":"REWORK","_number":1,"created":"2022-09-07 23:09:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/856348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/48/856348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/48/856348/1"}}},"commit":{"parents":[{"commit":"b88db4300afa919c5d977dace2fef59180f2aa8f","subject":"tests: Replace invalid UUIDs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b88db4300afa919c5d977dace2fef59180f2aa8f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 22:51:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 23:05:42.000000000","tz":60},"subject":"tests: Comment out prints","message":"tests: Comment out prints\n\nThese were polluting the output of the test runs and making it really\ndifficult to identify failures. Silence them. People can always\nuncomment the prints if they need them.\n\nChange-Id: I9aad5f3845f8abc64b1b2abc78b829929f73aafd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e9ae06acb0179042632857f5f82d9f2adf98332d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e9ae06acb0179042632857f5f82d9f2adf98332d"}]},"branch":"refs/heads/master"},"de46b3661b4875fa75cc3a98c59457612dd1cb33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-08 10:00:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/856348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/48/856348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/48/856348/2"}}},"commit":{"parents":[{"commit":"c816f67a64aa78201a429481693f1407f1a53056","subject":"tests: Replace invalid UUIDs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c816f67a64aa78201a429481693f1407f1a53056"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 22:51:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-08 09:59:51.000000000","tz":60},"subject":"tests: Comment out prints","message":"tests: Comment out prints\n\nThese were polluting the output of the test runs and making it really\ndifficult to identify failures. Silence them. People can always\nuncomment the prints if they need them.\n\nChange-Id: I9aad5f3845f8abc64b1b2abc78b829929f73aafd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/de46b3661b4875fa75cc3a98c59457612dd1cb33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/de46b3661b4875fa75cc3a98c59457612dd1cb33"}]},"branch":"refs/heads/master"},"5a644361b025cb5bcdcb7288c40e5553500b4f33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-17 09:34:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/856348/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/48/856348/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/48/856348/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/48/856348/3"}}},"commit":{"parents":[{"commit":"a7124dbdf46a51ed5a88b7005de3b8307c454973","subject":"tests: Replace invalid UUIDs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a7124dbdf46a51ed5a88b7005de3b8307c454973"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-09-07 22:51:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-04-17 09:32:58.000000000","tz":60},"subject":"tests: Comment out prints","message":"tests: Comment out prints\n\nThese were polluting the output of the test runs and making it really\ndifficult to identify failures. Silence them. People can always\nuncomment the prints if they need them.\n\nChange-Id: I9aad5f3845f8abc64b1b2abc78b829929f73aafd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5a644361b025cb5bcdcb7288c40e5553500b4f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5a644361b025cb5bcdcb7288c40e5553500b4f33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
