)]}'
{"id":"openstack%2Fironic~860198","triplet_id":"openstack%2Fironic~master~I4103152489bf61e2d614eaa297da858f7b2112a3","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-10-14 00:49:33.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-10-14 00:49:33.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I4103152489bf61e2d614eaa297da858f7b2112a3","subject":"Fix allocations default table type","status":"MERGED","created":"2022-10-03 23:34:30.000000000","updated":"2022-10-14 00:51:08.000000000","submitted":"2022-10-14 00:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"860198","meta_rev_id":"bad4bf8b5b5ba5d3dcb0ef0f09b9759c9ca9d443","_number":860198,"virtual_id_number":860198,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 00:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-10-13 21:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2022-10-13 22:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2022-10-13 22:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-03 23:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 00:19:38.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-10-04 00:28:41.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-10-04 00:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-04 15:39:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2022-10-06 22:30:55.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2022-10-13 13:05:31.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-13 22:04:16.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"c80257f9a0288acc0b76edc884f3c259eb856141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-03 23:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f2cd6d9ab3965cfc2b40c2207b3e8b92b157dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-03 23:34:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3e2a0dcc89c9d9533742e8ca6c89cfc6d3fe2ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 23:49:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9b7758e1ab40c8b2eea9311032288d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/368bfc622c2f4bc296c2302da9cf6562 : FAILURE in 10m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0419291402844d97a884922a6f03bdf9 : FAILURE in 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c53408e479ec36f37223d8da9b560d47a49281e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-10-04 00:19:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/51fb0b5/51fb0b5.html : FAILURE in 33m 07s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/2/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/1fd623c/1fd623c.html : FAILURE in 38m 53s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/b86e3b0/b86e3b0.html : FAILURE in 34m 45s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/cc21ad5/cc21ad5.html : FAILURE in 36m 22s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/2/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/bcbae2b/bcbae2b.html : FAILURE in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eecb0217b59410d3b183cb590405523e94a337f1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-04 00:28:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b92a11bf45bb46049aedaf5fe8a39dc7 : FAILURE in 26m 48s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2adc785676a646cb9d8d5c4549b9205c : FAILURE in 27m 19s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/342fad9af3514cf2b8d0b76f7dd16bfa : FAILURE in 26m 48s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3fe9833284f84019babcdabfd4940177 : FAILURE in 27m 28s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/44526d4c5a17416985b6524ce1071820 : FAILURE in 27m 14s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/86bf0118b7584b37bbade71d47320196 : FAILURE in 26m 35s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5fd8119d8f6f4021bb8aabce9037fe8b : FAILURE in 27m 18s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6eda7df94d1945489c6229a4a3ec3a0e : FAILURE in 23m 01s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0fb0aa89312f4212bf7dc1792b7ad45b : FAILURE in 22m 54s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/30440b9bb1e149b19b29d4b2e4781b08 : FAILURE in 22m 44s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7b73ab2dbd084fed906e9c64d35f749f : FAILURE in 22m 24s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"99457a0d60d555aa84d38c7ffb2b9af5f8b6d5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 00:39:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ff1fe2f251e4d6092c10ccea57895d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225d721aecb448c598f5e9ee82d4ecf5 : FAILURE in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ccd1c614f34714ae12b07706526824 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55789b7bb63741aa9aae0e92e8735389 : FAILURE in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba65dca069564adab8b8ef2ebd541a65 : FAILURE in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8413eac68dd44c28ab5c5029e59f666e : FAILURE in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e62348498347839c5dd88c20767b69 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78f00fa1e8a47a5938b65ceb1a98cc6 : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/30d23a59e057472bb12a347792116550 : FAILURE in 9m 40s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7c7a9218a68140479799729d0fbf1de0 : FAILURE in 8m 34s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/398434853da943ceb79357ed66a697ea : FAILURE in 21m 17s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/fdaf973bd1154f5bb3b553b4eaac0f4b : FAILURE in 20m 15s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/38faa92abde845b381efebe618d48988 : SUCCESS in 1h 03m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a6b4ab677f0748fe863f7d2e3a08fdca : FAILURE in 26m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/2dbfa5e3a3d442dc9b7fca93994d9c84 : FAILURE in 26m 57s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4f69c2101f4442ecbd90e59718960406 : FAILURE in 23m 35s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6e0c04dd888d43d487491fe0df5451b2 : FAILURE in 27m 12s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4d95bd3b887a435a925995fcb09d5dcb : FAILURE in 27m 53s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2bc35c89b9ad407181802bccc667a09a : FAILURE in 22m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/b494529748ab4f71aa25066559e76b43 : FAILURE in 28m 12s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/6804b4a5115f4573a7de87c5eceb8f4d : FAILURE in 31m 12s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/08496c5d1a1349c5a1563bb80950f72b : FAILURE in 22m 43s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/9e8e51f72d2a44aaa8c40ea0dc185a10 : FAILURE in 21m 27s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/7f0125b9f7f3432dbe76da6beb8152dd : FAILURE in 23m 17s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/484a65615c1946fba8c1448b6d52b051 : SUCCESS in 4m 26s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4ec0d7686aea4ef4a3e9c544a5fe78df : FAILURE in 28m 01s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5ea2025ce7674415ac4c7002bd8cb06c : FAILURE in 32m 27s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/cf098f86b06d4683a452d815b758ac98 : FAILURE in 27m 25s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d5cf33bf226347c7b3f55712d178cd67 : FAILURE in 23m 44s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/20c251840e3447f0a4f84a7838afa36b : FAILURE in 25m 06s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/cb723cdf805447d5acec983d242fc471 : FAILURE in 25m 28s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ee75a080854e48a78668c8ef6c907a : SUCCESS in 33m 24s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/2f364916a57945b784d3983c98e2a66b : SUCCESS in 35m 21s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/b2bade0c844f41aea354161e3f897e2e : SUCCESS in 43m 17s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/a1f5f374efe14662b449d34d0f4a62ce : SUCCESS in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09be49b7f287b5e26a752168a9c12c601bc6078e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-04 13:09:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0518c761329a2ccb8acf8a2c0d6df4b1e0c58dd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:39:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed7c60fa2354bf0bd332b9242e0d661\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd6560f31f97490d83d985af070cf69d : SUCCESS in 15m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5107cb24800a4f84b033a99f03fd004f : SUCCESS in 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ad0732fb293b5e75e509966be04e6f5a8e7b06e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-10-04 14:20:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/f72671c/f72671c.html : FAILURE in 1h 04m 44s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/3/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/c61d03b/c61d03b.html : SUCCESS in 45m 12s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/105fe1f/105fe1f.html : SUCCESS in 47m 27s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/b46e949/b46e949.html : SUCCESS in 48m 19s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/3/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/6865a81/6865a81.html : SUCCESS in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94231177a6576db381cddc792359e242be3ecc10","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-04 15:03:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b20c4bb1e0af4327aa55809078986d7b : SUCCESS in 1h 00m 56s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e39d1c5e6fa5453aa6c960806a8b6394 : SUCCESS in 57m 06s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/33c9382621fa49a0a48da027874c5296 : SUCCESS in 56m 05s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a974ae2c0fd04c449cb032586f90078b : SUCCESS in 58m 37s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c2ef9a62b99f49ff957bbd98e3e917f5 : SUCCESS in 56m 22s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6cee75b9ce1b4afcb3d41da20fc90f5d : SUCCESS in 57m 24s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fb0e107557f8450189af4c20b27b4fee : SUCCESS in 55m 20s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/75bc2242411340c9b8dca8c0c520d033 : SUCCESS in 48m 46s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/aac34e93570c49ccaa0cb0ccb80e4fd4 : SUCCESS in 51m 01s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/06f26bb34f9a497da0052f58fab554f9 : SUCCESS in 49m 22s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1d6eec4e687b4f578d331ddf5c0f1ed7 : SUCCESS in 52m 24s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"d04ca941de64c42b14fd64ec555a348527a8dab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 15:07:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a7094456db745ebb54c70a29779c278\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f11e32ae88346969ebc17970d517fe3 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aed9d3dd07e404cb4f985d2ee4efd5f : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8ec2226032749d1a0fe2b82c51c78dc : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2599a973eeec47bfa617260d13c420ac : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8da00888ccd248579d768987496a76e0 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a493b231f0ab437d8f7329e3d13b6125 : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a262845cbb064cb68767a5088c94c140 : SUCCESS in 8m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f8e3b92f866b43fa8aaf8a7bbf050198 : SUCCESS in 11m 08s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9a5d6e277ff047c2bf2528e526c8667b : SUCCESS in 9m 06s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3bacc704786941ca9bcb17b831b76d43 : SUCCESS in 33m 55s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/569710b982954565a0067a9e60c6a0f8 : SUCCESS in 29m 30s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5f890ece22024f3c898de87f227e73d2 : SUCCESS in 1h 09m 15s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/44bbe37daf8c4fe3b6db2ba9917e4193 : SUCCESS in 1h 13m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/02b313ab0df6486788e441d0c2d87946 : SUCCESS in 1h 15m 38s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8e60e459c9014a31a5dcf047c40aff3b : SUCCESS in 52m 16s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a35464933f124ead947d97b727adda6c : SUCCESS in 55m 50s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/61a23eca86b040bfb740d16bc8c5ae3c : SUCCESS in 41m 39s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b941951e272f474284f3daf674dee449 : SUCCESS in 59m 07s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/75fbd4ebd0b141769df1c8ea59c4983f : SUCCESS in 1h 54m 30s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ee976c46846c45a48127ffd6f29c9c46 : SUCCESS in 1h 04m 00s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2c822c4e426e47ff94e3c384672aad53 : SUCCESS in 55m 20s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/fb9265490cf54587b1bf4bdfe720f8ec : SUCCESS in 53m 06s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/c263cc2b00f34822bc534b13b7d01a9f : SUCCESS in 1h 21m 20s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/93625689e8304ea8a07121fe50c10874 : SUCCESS in 3m 57s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0da1351b24634429a852dba5cbf2b2cd : SUCCESS in 54m 29s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/ec7c7f4cd8764d55ba90b832747263d0 : FAILURE in 1h 12m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/205f5aa0e78b4b4bae735007493c5d6c : SUCCESS in 54m 13s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/25213a4204bf407397aade8bdc3b3398 : SUCCESS in 59m 50s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/8baea1f4493a4d4885616ce67482aaba : FAILURE in 31m 48s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/9f37008c73dc4301a386da678d18ee3f : SUCCESS in 1h 07m 21s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f7c85182f1e4fb593300aaaeafbc99b : SUCCESS in 39m 56s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/78a38fcdd2ce46148f5917bc81db0391 : SUCCESS in 40m 49s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/7dac15dfe2c44d90adf34fa9f018d33d : SUCCESS in 1h 04m 42s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/1a97d27ac4d048f09d1acf359f9835dd : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9869385be90f20e8431172707eee30b24c042974","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-10-04 15:39:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c424bddee5f480c876105cb5a95d74e86dbf0d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-04 17:19:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"da39ecb02e8a44d9beec11c3394595161ccdd392","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 18:02:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13574f29b4714fc89f6d915bf255726f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6aee5bcdc7034802b67f0d1c18df8af4 : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0422a1e8781485d86b027694552c10c : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1899b6895ffa990f8e1065c798a2216400ab7a3d","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-10-04 18:29:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/36097ed/36097ed.html : FAILURE in 1h 02m 49s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/4/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/9b8a387/9b8a387.html : SUCCESS in 45m 14s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/bed0733/bed0733.html : SUCCESS in 44m 24s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/6bb0806/6bb0806.html : SUCCESS in 48m 25s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/860198/4/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/6f7c88b/6f7c88b.html : SUCCESS in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99df404bc628fda3f970cf4622eba967fa5197d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 19:00:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d94a9670d67f486083582144c8b43bad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab275244fbd8466c84f509f6c96caf88 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adafbaa53ed34112a04cfb44f4970491 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/109f5d3b2f6e48c4912d6ceb1e6dc2c0 : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/890183de6cbd4338b8ac4f72341f83c8 : SUCCESS in 11m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b60e1f59b1b44949096bc2550b693d7 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ce94321bfa43dc8b0aaa9de83530c8 : SUCCESS in 14m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96513300ddc04f5a8c3c750ef164c5fe : SUCCESS in 9m 47s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f37c094e11864639b78a9d303ba9764c : SUCCESS in 9m 54s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6ebaa83039fd4fabab5057392dd542d8 : SUCCESS in 8m 07s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a23954a531f44ffda73ba075db7c1148 : SUCCESS in 28m 33s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/7dfd168eb2454c0f9a05682849dbe751 : SUCCESS in 30m 14s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ac32bcef213246979fb359c1937208c0 : SUCCESS in 1h 13m 14s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/df1eefb6f08947a1b7d9bc77d1084304 : SUCCESS in 1h 01m 15s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0f67d51f3fcf408da7e1e77ea644feca : SUCCESS in 53m 44s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f1b725ac2ea246d396894774de1244f3 : SUCCESS in 42m 47s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f21037b88e754b83a9fb8ce00987ef73 : SUCCESS in 1h 03m 04s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8d9aeef2761b4c298ec7b18fcc19c2e4 : SUCCESS in 42m 04s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/adf173f42072427d8ebbab42b6f09891 : SUCCESS in 1h 00m 45s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/3a078c0e896d494a871724d6dc5c60b5 : SUCCESS in 1h 32m 51s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4a50435d795844f4a7a191b79af36b53 : SUCCESS in 1h 14m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c48bed5c0ffb4346a520cc5ca112e2d5 : SUCCESS in 39m 32s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/021bb7e19408492392b14d931ae79273 : SUCCESS in 50m 46s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/3d70cb5e6c934ccd8116771c2ec73da4 : SUCCESS in 1h 19m 24s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5dd4f41c53b14240ac8e9a91b8470c0e : SUCCESS in 6m 04s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/706f442509a74b15945406bc82ae5c56 : SUCCESS in 53m 15s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/06461a8b0f944ab0857fb7e751d76c79 : FAILURE in 54m 20s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/77f126ab349c4c1492d835b7d743dd63 : SUCCESS in 49m 48s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/1e9c15c14179461eb5d16eddb9f55a07 : FAILURE in 1h 19m 49s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/fdd4a6b356b44a3985648036ab5cf6b7 : FAILURE in 31m 47s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/42c289e6bbbe4741b74e8e1f5afbec17 : SUCCESS in 55m 59s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3bd80bbb35e45c8a8dc795d3104891d : SUCCESS in 35m 13s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/780f57e4c2264788b96c3eca50460703 : SUCCESS in 34m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/76d6e691183c49eb89c997c57aafd2f2 : SUCCESS in 1h 01m 34s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/f769a881be7946b897788e5763003310 : SUCCESS in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"abf5d5cfe2cfd3b0bf7df4a55500351862e55eae","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-04 20:18:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d5a0fe0ed66743fab61c12d4b27ce6a2 : SUCCESS in 57m 46s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/aa7c8973a0874b9d975f75c7e065b9a8 : SUCCESS in 51m 51s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4bde92c83a0740be95ab445e9d9d0579 : SUCCESS in 52m 26s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7ed783520ba84a47a6d475fff0b9f537 : FAILURE in 26m 41s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6aa37c271edf4a829d30ed9e89cf0d96 : SUCCESS in 51m 17s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d9ba7fa594ca4b6cb581f44ce281090b : SUCCESS in 53m 25s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9d9a9f2007614891af205ef9a1a3dd9d : SUCCESS in 52m 03s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7f661a63763a402aad4b92874f596691 : FAILURE in 26m 33s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4fc2958a8ad446abb57b0b0470f8b63e : SUCCESS in 54m 31s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5e5be7de428c44ea87a0b352d70a3dac : SUCCESS in 53m 41s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/94ad9b7f51c9476db7b128807e36396f : SUCCESS in 56m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"684042131301ab7e3633e4455b0284a92f94d407","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-04 21:32:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a170ed4b77f6888b96cdb19dffe047f5d5b3a346","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-05 13:28:54.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":4},{"id":"fd845bbcf7491b8fecee71997422c4dbaa25f013","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-06 22:30:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d5700385ddeee0484f46c20bf58890c44962ea2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-09 19:25:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d6fc248dad7d730afd1ce9021e49d45cbe0203dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-12 20:58:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"67691cb89be243f32d41d9094c30e0dff5860856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 21:19:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5942edb985864a91ab88cdfd58feabd6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0db6c9627a24445b6a6e320aa22f68a : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d2c71ca2d4942b39c6f7f6692ee4f09 : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3cfe4d9ff492ea665c053bba7ca29fb7cd01985e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-12 21:35:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2531ceca9798e4d149263f4b5d479b00890a8b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 21:48:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9074d814ea648d38249c2fe02087764\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcffe852e16b4b758ca9445a00aa8935 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c866b16cc786401e997bd4d62a37e085 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c749a4f1d3fa4bd6a49684beb85d2ea1 : FAILURE in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7168e1a7d32a4b9581f083c9e8ae43b2 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea18657d8e8b49e1a1bdd175b5d48f8c : FAILURE in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b8d7f4e3d24e5f8ac51f106a532da3 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cffa35d9dffe41aa98b610818e74e805 : SUCCESS in 7m 34s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2600a0fba5764ede8a38bb17a4a5aecc : FAILURE in 10m 20s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/aa167b5078c94eec9efee089c07ad722 : SUCCESS in 6m 44s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/545584c4677e4d868eb37f32f2868cea : FAILURE in 16m 40s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/df8d8243ef694c658d5f5eb7b2f5351d : FAILURE in 13m 28s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5d3781da2dae4896bc91ca9e7450f402 : SUCCESS in 47m 57s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/9cbbec7ead2041edbf4ca472c0a1e90c : FAILURE in 21m 41s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/aa6e145ba8794a93ae05a1f3988790d8 : FAILURE in 21m 35s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a42daa4779c8478aa3324e6335db7bb5 : FAILURE in 18m 14s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1e9a779603414f0fa707c6878175ebff : FAILURE in 21m 02s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/af8925d441fe4f77a84006c2a7c3f7b7 : FAILURE in 22m 55s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4e29dfd618f9464ab84d3092dcec0d71 : FAILURE in 18m 14s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e7718a0c49f54b04a095151eef9a88f2 : FAILURE in 29m 34s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8d9e2b2c03f94711a6143252279abd85 : FAILURE in 21m 11s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c7a9480ae9fd4a85b06bd6aaf0877a99 : FAILURE in 24m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/68b416113da24da6bcd1cce4fac6d752 : FAILURE in 18m 25s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/5eee607a0d03451a8d0d7d07f1ceb175 : FAILURE in 19m 13s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/cb16503462ea4ae49921823088548d05 : SUCCESS in 4m 36s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/77797701b51a4399bd8799e222767f29 : FAILURE in 22m 46s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/df7b2b1800b24841b03cba11768c15b2 : FAILURE in 22m 42s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f44857c40f9143c58d82e41bdef86f54 : FAILURE in 24m 56s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c28b00295aa94b41b52cab43300f338c : FAILURE in 24m 58s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/282cf9ac4e4446aeb5ec14f076d4ebc6 : FAILURE in 16m 41s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1198fa6d5e394754923d11f2ef10de03 : FAILURE in 23m 14s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/471eef24fc5a49e891a659a6cb868b64 : SUCCESS in 38m 15s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/6146ee2f00984672a65ca4798e8b3722 : SUCCESS in 33m 44s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/668799c109a04169b4d2ae3e66b10c97 : SUCCESS in 41m 39s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/4de826d7dc9348b5abeb5bfeee5c6de1 : SUCCESS in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"81511b5af7a285a3d293b17caa7343fa24e3899d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-12 21:53:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4fa7d15cbb954a258f8e163e520968f8 : FAILURE in 25m 28s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7889b8dc3e9743b3b7a934e3686fd4ed : FAILURE in 25m 10s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b7f1e51509df45aa959d55806e8f02d5 : FAILURE in 25m 22s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d7df5b57f9b54a6cb1fd6ca888f66690 : FAILURE in 24m 47s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d8a2b1671af64dc9840b88e1e9ff4388 : FAILURE in 24m 43s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1cd73cf129e34bd4a56b7abb080d6b89 : FAILURE in 24m 18s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8d59815c621541b1a9882550c26493c0 : FAILURE in 24m 20s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0f7fb2fdc7b64aae886e2d630d7e7225 : FAILURE in 25m 47s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bdce623f74e842a9abeee559d0e4b44d : FAILURE in 25m 25s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f9c36367dbaf4163bc6223f63ef7cfbd : FAILURE in 25m 26s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d5e8004f016f4ed38ac581ba7ff942cb : FAILURE in 25m 09s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"478117d149046addd682993d7bc51da995283eef","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-10-13 13:05:31.000000000","message":"Patch Set 5:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with FIPS enabled \u0026 the other is FIPS disabled, both with CentOS 9 Stream\nand PXE boot for Ironic node boot, then we run following Tempest tests in both environments \n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-fips http://openstackci.jp.fujitsu.com/irmcci/98/860198/5/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-fips/7537065/ : FAILURE in 50m 44s\n- fujitsu-irmc-ci-job-nofips http://openstackci.jp.fujitsu.com/irmcci/98/860198/5/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-nofips/02b5ed2/ : FAILURE in 44m 46s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":5},{"id":"7b6c7089dc319a43eb3c5d13b86da4838f75417f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-10-13 21:21:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6c607f19232a5915cb674bd409e338f8a43ce311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:39:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd1bb70dc1c4490b37a0639882b23d2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f65fbdcc0a24e7db178e84d220eac1d : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/913464e944414f15b45c7e56e2591d11 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e9c9faaf940f2ca193efb85a0949e219bbf46f8","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-13 22:04:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"57cafb7e606ec289b501c17f75283fdedf2ab53a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-13 22:55:37.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3b763e82e2014f9790c10a0323c2ce53 : FAILURE in 26m 50s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3380a12fb63a43f4abb91c899cdd3781 : FAILURE in 27m 10s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4a2a60ed155d4c4593556026e39b9b46 : FAILURE in 26m 44s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/61de3dbc06b04b6183838621d2bed535 : FAILURE in 26m 48s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/de92c30407c94b3e82557d09a3e4048a : FAILURE in 26m 58s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/07122ebe2692460eb8ebe57c4118a31c : FAILURE in 27m 10s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bcab6c98eaf745ddb2c6ec429b4c7990 : SUCCESS in 52m 35s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2da7c01c6e884c03bbdf910a23aab864 : FAILURE in 27m 11s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c6f4fd5e95434c77bdbabff4dcf7038f : FAILURE in 26m 56s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cc9701a9b2dc416dbf429fe8d06ff139 : FAILURE in 26m 38s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/132d6577e8a84ee2abd1dcfdaf30e957 : SUCCESS in 1h 05m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"6c0b231ed0098da39a2a7f4711637d9d372bd1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 23:00:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b979caef904dd7a51b0041f6a08d7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da429edcdbe34bf2a160b9782839ba05 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d11588ab9c490893422436ad16e4c7 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aff3d350ad2a4cd98a5aa41a0ca4c2cf : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9861b06847e41d8b7d3e6a914b7526d : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f040a2309e04ce6a4e6ab4923f73a6f : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fab7a6a0c9242978476946d72ca3731 : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed4672bbc0414b8b890cdb66e1547a4f : SUCCESS in 7m 31s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/dc86329f286b43148805206ebc4c779c : SUCCESS in 9m 43s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e63c686f39f14ab3a412560675cd27c9 : SUCCESS in 7m 44s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/2dab48a1a44b4598ab512a545ac73596 : SUCCESS in 17m 47s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/64367ebcfa7944edaad8aa8f6f906876 : SUCCESS in 21m 34s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6aae1247e5974b30bf286a0d3b921d57 : SUCCESS in 1h 08m 39s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/89515d46725744539d05d76ecf74f5e2 : SUCCESS in 42m 37s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b8332634247b4c48bbeaab192c9daf05 : SUCCESS in 49m 38s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5d3051471334474997f809caee1aa2f9 : SUCCESS in 34m 15s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e3b7fb4b188e46d1bc456d933941f1bf : SUCCESS in 52m 41s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/4715bbd51d9049558ff026c7dda08dd9 : SUCCESS in 31m 58s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2a45552ffb5947d69ac4e72e8dd345e9 : SUCCESS in 47m 24s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/47e5f74d262649c0bb6ab354d9603800 : SUCCESS in 1h 33m 21s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f00a037761d14e89be8f407c60a91eae : SUCCESS in 51m 39s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/a0e565eff07246ee9fb026b6154fd501 : SUCCESS in 32m 25s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3835f82ecbeb43ff866d94608e1dbb41 : SUCCESS in 51m 39s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/fc0dea7826814dd68f6614e60fcdb3a0 : SUCCESS in 1h 09m 52s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/3b68497ef2824ab9b4e51e04f8bd78b9 : SUCCESS in 4m 27s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/5a867c1b7b4a4ac3b264c5e617aa2871 : SUCCESS in 54m 40s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/73068c1627b84d4c91aa407c8c414f12 : FAILURE in 44m 15s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/c8d9590010b74dbaa6ad65259a88ff0f : SUCCESS in 43m 10s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/3f7b4c689ce2446c9c4d2565cd34965d : FAILURE in 1h 05m 00s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/e7e67a84a26547298e77c6cd2ad18f3e : SUCCESS in 33m 30s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/efcb49c1fa4d4b258c017fb7690ddf91 : SUCCESS in 51m 48s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3449b95ae7a04d349c992d58a88f0dfa : SUCCESS in 31m 49s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/dca28b6bb5fb472f8499dbe5c7148c8b : SUCCESS in 50m 37s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/173beb1b12c44aa98aadc49d21742746 : SUCCESS in 43m 58s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/dcf437c7941f44ab915c13a3b362dfbe : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4ab4e21e4ec6be28f259acd57f2c1f9179dbabdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 23:00:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2bcb84c067da431233db729cc095c84f3c095635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 00:49:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa581feb2b84ca0b1404deb0fddd1d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af573b256b5f412f80fe42b35c9562d2 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf8f8f05148a46fc9357e7b8f2dc1d8b : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85e1c9dfffd3463e950084cf675ee1a9 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6755101736f142c8a2da3d6e8136563b : SUCCESS in 13m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce65448a6db54ab0a630a9e1bf1558f3 : SUCCESS in 6m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/be06fcb050c04b13acaf12e3f21d6403 : SUCCESS in 9m 04s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/03ee33521a4049e0afab54e2c13be947 : SUCCESS in 22m 02s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e6a5c1d1ba8b44c5beee73309adda088 : SUCCESS in 21m 48s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/71a538849dbe4f89bd70252e6665a6f5 : SUCCESS in 51m 57s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7ee13fa254b14af3ba54ef341074c180 : SUCCESS in 51m 22s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/551ad521b2224a188d4c34bf7bbb8750 : SUCCESS in 46m 28s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ad1ebc88686c4a189ea784f3c4722fb7 : SUCCESS in 32m 23s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f27c3beb34b042ac8c1f6335a6c36c27 : SUCCESS in 47m 17s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/29610f8105734c7c84a197256f15d8ab : SUCCESS in 34m 43s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/45a804ebe8f149648f13f62a1a2562fc : SUCCESS in 40m 27s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/921fe274fa994b258c781fdad5a41d76 : SUCCESS in 56m 43s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/98c177c713b8427f9aebf276cfd56203 : SUCCESS in 33m 51s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/56f076b35b794ad8bddde5111b1d161f : SUCCESS in 49m 01s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/4fcd6b3ac52a487b81acccac745b5edf : SUCCESS in 1h 44m 07s","accounts_in_message":[],"_revision_number":6},{"id":"75058fb9135ddff7ac8e446ffd1e9bf8c295e8bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 00:49:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bad4bf8b5b5ba5d3dcb0ef0f09b9759c9ca9d443","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 00:51:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a169cafefb49e08acda69c15837b16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4da324d33814663873bf444533ddd44 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d77f9aece224a80b2fc50875d0948fe : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1435a15ce3013da0a3138e1c5ab6ac523c382bb8","revisions":{"704d7e346f6b3b6f8a863db184412d6eab5be818":{"kind":"REWORK","_number":1,"created":"2022-10-03 23:34:30.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/1"}}},"commit":{"parents":[{"commit":"22642431168b87833b14e9a0d517da36498c0094","subject":"Change allocations name length check to an upgrade failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/22642431168b87833b14e9a0d517da36498c0094"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:30:10.000000000","tz":-420},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/704d7e346f6b3b6f8a863db184412d6eab5be818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/704d7e346f6b3b6f8a863db184412d6eab5be818"}]},"branch":"refs/heads/master"},"7369014387a1802c427eaab9f375e305aa007066":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-03 23:34:41.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/2"}}},"commit":{"parents":[{"commit":"3b1504f4abeac357298ea967d63787f53aa977f6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/3b1504f4abeac357298ea967d63787f53aa977f6"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:34:41.000000000","tz":0},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7369014387a1802c427eaab9f375e305aa007066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7369014387a1802c427eaab9f375e305aa007066"}]},"branch":"refs/heads/master","description":"Rebase"},"ea84674f63a0bc156e7d4d5e5dd6e0bc816a3532":{"kind":"REWORK","_number":3,"created":"2022-10-04 13:09:28.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/3"}}},"commit":{"parents":[{"commit":"fdfbe6b6810a3962f503aebdb4fe1eccd5d98ffb","subject":"Add missing space to log message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fdfbe6b6810a3962f503aebdb4fe1eccd5d98ffb"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-04 13:09:22.000000000","tz":-420},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ea84674f63a0bc156e7d4d5e5dd6e0bc816a3532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ea84674f63a0bc156e7d4d5e5dd6e0bc816a3532"}]},"branch":"refs/heads/master"},"df490e61bffcfc1147fdd7ce31299ae4ed0d45c2":{"kind":"REWORK","_number":4,"created":"2022-10-04 17:19:57.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/4"}}},"commit":{"parents":[{"commit":"fdfbe6b6810a3962f503aebdb4fe1eccd5d98ffb","subject":"Add missing space to log message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fdfbe6b6810a3962f503aebdb4fe1eccd5d98ffb"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-04 17:19:56.000000000","tz":0},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df490e61bffcfc1147fdd7ce31299ae4ed0d45c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df490e61bffcfc1147fdd7ce31299ae4ed0d45c2"}]},"branch":"refs/heads/master"},"df017c7982d4921419132ac3ea36cf154fac90c6":{"kind":"REWORK","_number":5,"created":"2022-10-12 20:58:57.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/5"}}},"commit":{"parents":[{"commit":"38831529900448be54b4760782f2438d290d1d69","subject":"Add upgrade check warning for allocations db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/38831529900448be54b4760782f2438d290d1d69"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-12 20:51:51.000000000","tz":-420},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df017c7982d4921419132ac3ea36cf154fac90c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/df017c7982d4921419132ac3ea36cf154fac90c6"}]},"branch":"refs/heads/master"},"1435a15ce3013da0a3138e1c5ab6ac523c382bb8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-10-13 21:21:24.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/860198/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/98/860198/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/98/860198/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/98/860198/6"}}},"commit":{"parents":[{"commit":"9344eb22d1462457fb5b521a19856f9a6cb1088e","subject":"Add upgrade check warning for allocations db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9344eb22d1462457fb5b521a19856f9a6cb1088e"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-03 23:06:37.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-10-13 21:21:24.000000000","tz":0},"subject":"Fix allocations default table type","message":"Fix allocations default table type\n\nIn trying to figure out why I was unable to run\nall of the test_migrations tests, I realized we need\nto fix and clean up our unicode declarations.\n\nSpecifically, the way I found this was my local mysql\ninstall was defaulted to using 4 Byte Unicode characters,\nhowever some of our fields are 255 characters, which do not\nfit inside of InnoDB tables.\n\nThey do, however fit with the \"utf8\" storage alias, which is\npresently short for UTF8MB3, as opposed to UTF8MB4 which is\nwhat my local database server was configured for. Because this\nwas in opportunistic tests, I wasn\u0027t able to really sort out\nwhat was going on and thought we needed to shorten the fields.\n\nIn reality, it turns out we never defined the allocations\ntable to use UTF8 and Innodb for storage.\n\nStorage engine wise, this is not a big deal, but may mean a\nDBA will one day need to dump and reload the allocation table\nof a deployment.\n\nCharacter set wise... It is not great, but there is not a good\nway for us to do this programatically. In my opinion, the chance\nof an issue being encountered by an operator is unlikely, which\nout weighs the risk and impact of dumping the entire table,\ndeleting the table, recreating the table with the updated schema\nand then repopulating the entries. Of course, if operators are not\nusing allocations, then it really doesn\u0027t matter for them.\n\nAlong the way, I discovered we had used the \"UTF8\" type alias,\nwhich may change one day, which would break Ironic. As such,\nI\u0027ve also updated the definitions used to create databases\nand updated our documentation.\n\nRecommended reading:\nhttps://docs.sqlalchemy.org/en/14/dialects/mysql.html#unicode\nhttps://dev.mysql.com/doc/refman/8.0/en/charset-unicode-utf8mb4.html\n\nStory: 2010348\nTask: 46492\n\nChange-Id: I4103152489bf61e2d614eaa297da858f7b2112a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1435a15ce3013da0a3138e1c5ab6ac523c382bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1435a15ce3013da0a3138e1c5ab6ac523c382bb8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
