)]}'
{"id":"openstack%2Fironic~866656","triplet_id":"openstack%2Fironic~bugfix%2F21.0~Ic5a235785a1a2bb37fef38bd3a86f40125acb3d9","project":"openstack/ironic","branch":"bugfix/21.0","topic":"align_with_ironic_boot_mode-bugfix/21.0","attention_set":{},"removed_from_attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2023-02-20 15:37:09.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-02-20 15:37:09.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ic5a235785a1a2bb37fef38bd3a86f40125acb3d9","subject":"Align iRMC driver with Ironic\u0027s default boot_mode","status":"MERGED","created":"2022-12-06 15:56:20.000000000","updated":"2023-11-09 12:55:12.000000000","submitted":"2023-02-20 15:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866656-align_with_ironic_boot_mode-bugfix/21.0","cherry_pick_of_change":862552,"cherry_pick_of_patch_set":5,"meta_rev_id":"24c1b73fa743e743ca42ccb0bf45a711058269d2","_number":866656,"virtual_id_number":866656,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 15:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-20 14:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2023-11-09 12:55:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-20 14:07:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2023-11-09 12:55:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 15:56:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"state":"CC"},{"updated":"2022-12-06 16:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 18:09:13.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-12-06 22:38:11.000000000","updated_by":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-20 14:07:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2023-11-09 12:55:12.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"ecef859477a04eae4d4a1e9dca95731e0d98ab0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-06 15:56:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c8db76b9abdf082263e023f78e4a70fe1841f504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 16:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4465e185ef54447388c1a12a609eb7b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66950bc8017a4ebfa190f603d670b744 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b23804460f4607a0b1ebe7aeb7537a : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83f9acc8c31f4a5e98adb432f341cf91 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9d0abd4ed8a46379188459546278ab4 : SUCCESS in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e97263c96dd74bb8805f0a074d718e25 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fc592135f274d87bd0c002710336f39 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91b7098864c3472b8ea27f0739692528 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2606c9c2104e6da735774b2fa1d012 : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b55a2c7b1da45cbb3fb346c8dcbfcf6 : SUCCESS in 7m 08s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/2085af310d98434fa4ea60734b2565ac : SUCCESS in 10m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/75c540a959f1416696199371b7736239 : FAILURE in 5m 29s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/78b721967d9d431ba04629445787ea6f : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92a1c4c9bd50e36d68bcb521dbe78554fecc5260","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-12-06 18:09:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b0668db5d6b94e21a8446ea582175988 : FAILURE in 26m 10s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2f0b047d41f842669698e33d9d2cfe09 : SUCCESS in 52m 17s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1638cb150b05466ab93fd0568c478a92 : FAILURE in 25m 47s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4579b3f30b5e43e48de192f4c8db3d37 : FAILURE in 26m 02s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/af277bb368cf4d3f8b880dd644f16fe4 : SUCCESS in 50m 10s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6c02594e237d44eaafeabc0d285c4a6e : FAILURE in 27m 49s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2ac7e461a41144ecb599874cfb401b76 : FAILURE in 27m 46s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c7a640cf2c534fab87113df6f99e04c7 : FAILURE in 27m 42s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2f8fda788f7b4267886275cef21d8f2d : FAILURE in 27m 06s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/46aa5b4daf3b4bfea1ef321188074756 : SUCCESS in 52m 12s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d21f18c0ec334cfc863d18a973ea2f74 : SUCCESS in 58m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"4ca712e5e623ce4770a99cb4bf0494400e0a6296","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2022-12-06 22:38:11.000000000","message":"Patch Set 1:\n\nTempest test failed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with FIPS enabled \u0026 the other is FIPS disabled, both with CentOS 9 Stream\nand PXE boot for Ironic node boot, then we run following Tempest tests in both environments \n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-fips http://openstackci.jp.fujitsu.com/irmcci/56/866656/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-fips/d9f977b/ : FAILURE in 42m 39s\n- fujitsu-irmc-ci-job-nofips http://openstackci.jp.fujitsu.com/irmcci/56/866656/1/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-nofips/a657f45/ : SUCCESS in 1h 15m 39s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":1},{"id":"5bf1c7737153fdd8d7fb2eaa75d0fcdea294baf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-12-07 22:45:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29cc1009c388037b86f506f211967bed04f4098b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 23:06:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5e48aa2d0d45ef9a6a3a4421f65a6e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6ce7384ac9e4703b603c582bb48b51f : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6b4a95d0c64999958da68899ee1549 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e978152f9a4acfb7e81f6f2fce39a7 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706c0a7263324ceb961a147bfda291b2 : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299fa198d088430989172e463ea571d0 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb60016b304044c7a02afdd87b57b70a : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c72258d11e84bf89063b35208b1935d : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca869d734354b388498d90ed301fb30 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23a8f2e981534ed894d76b2ade8d844f : SUCCESS in 5m 38s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/b5a15fa7553d470bacbed0a963b9741d : SUCCESS in 10m 53s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/29050add1b99440cab3cbfa853f2a820 : FAILURE in 3m 29s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/9a339a029d294e08b4a3f106b0c1bfab : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7355c448ea0b76508fbb031ebf9d0ecf244d0356","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-12-08 03:53:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- R650-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f55774d7b5fe49fc94b6193ddc995c00 : FAILURE in 26m 26s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f4589ff2de464a4a9ab2cfd7c4717151 : FAILURE in 26m 08s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e8a212fd10e74f62b99fcbd1130f7788 : FAILURE in 26m 51s\n- XR12-uefi-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cb677bcadb924ef49cf9c44cce2e228f : FAILURE in 26m 01s\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6a245b0c1f3a48988e3e771edf625795 : FAILURE in 25m 51s\n- 14G-uefi-ipmi https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c02826815983467097c441209b7c4101 : FAILURE in 29m 09s\n- XE2420-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4e47ecc214b441268459ad2c79bf06fb : FAILURE in 25m 43s\n- R7515-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/372ce37e268e4ed6aeadd77797994864 : FAILURE in 30m 54s\n- 14G-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/22d9f0cb2e624d7199feba3553202152 : FAILURE in 30m 52s\n- 14G-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7d5eeca032024ca5847d55caa13eb89b : FAILURE in 26m 09s\n- R6515-bios-idrac https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2ffea8f627394762b3a7d28d717d3a42 : FAILURE in 29m 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"8f45c0329c1d1a3fdae35865f1344254e7c110a4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-16 11:07:29.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"a55f7550a89ad37c5f395c072fc670a5494ed687","author":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"date":"2023-01-20 04:58:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f58d28d057ddc77ca28232d60ea35c1211f21cc7","author":{"_account_id":19003,"name":"Fujitsu iRMC CI","email":"fj-lsoft-irmcci@dl.jp.fujitsu.com","username":"Fujitsu_iRMC_CI","tags":["SERVICE_USER"]},"date":"2023-01-20 06:22:12.000000000","message":"Patch Set 2:\n\nTempest test passed in Fujitsu iRMC CI environment.\n\nIn this CI, we prepare 2 environments, one with FIPS enabled \u0026 the other is FIPS disabled, both with CentOS 9 Stream\nand PXE boot for Ironic node boot, then we run following Tempest tests in both environments \n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_partition_image\n  - ironic_tempest_plugin.tests.scenario.test_baremetal_basic_ops.BaremetalBasicOps.test_baremetal_server_ops_wholedisk_image\n\nIt\u0027ll take few minutes till log appears in following URL.\n\n\n- fujitsu-irmc-ci-job-fips http://openstackci.jp.fujitsu.com/irmcci/56/866656/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-fips/a02a183/ : SUCCESS in 1h 21m 29s\n- fujitsu-irmc-ci-job-nofips http://openstackci.jp.fujitsu.com/irmcci/56/866656/2/Fujitsu-iRMC-CI/fujitsu-irmc-ci-job-nofips/828135e/ : SUCCESS in 1h 18m 43s\n\nAbout this CI test, see https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_iRMC_CI","accounts_in_message":[],"_revision_number":2},{"id":"2fda12b455c294a312a26141897357c3b854ddd2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-02-20 14:07:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e7ef4af3c40906a12734b542a8d1a6b311ca6cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 14:08:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38310cd17980c00d36a3b3e3ba657132f9293167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:37:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847e17630e634381857e1c49a8195c44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f9adb14b78545e0b9026616fdd3f8bf : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0341a1e93fd44a4a258441b1f71bc36 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4c0dec441534e84a6101d9ff332d54f : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e42ab1dcf0487ebb3bd7e3f3485fed : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c564d1a38ea24fadb44082ba9eb304af : SUCCESS in 13m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d0ff11a030984b2db75c0e2e93790ad4 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4a4f1eab9f7bd54dc0aece40613b2ddf7965c76e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:37:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29e284ad62e13bca26da7caa4c960285b3b4e97f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:38:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/698aeff05fa849759390c9ea12586060\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b4f9d144364c829e88a6f2ea5d4660 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2},{"id":"24c1b73fa743e743ca42ccb0bf45a711058269d2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-11-09 12:55:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29546c18dccf19d80e2fac866677538d641f8be6","revisions":{"95341123a33f0bd58e56cff2fe494f39da01e947":{"kind":"REWORK","_number":1,"created":"2022-12-06 15:56:20.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/56/866656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/56/866656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/56/866656/1"}}},"commit":{"parents":[{"commit":"fe64d5e1ecd3bdeab25232d4ab06ef984753547f","subject":"Merge \"Add support auth protocols for iRMC\" into bugfix/21.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fe64d5e1ecd3bdeab25232d4ab06ef984753547f"}]}],"author":{"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","date":"2022-10-25 05:14:44.000000000","tz":-240},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-06 15:56:20.000000000","tz":0},"subject":"Align iRMC driver with Ironic\u0027s default boot_mode","message":"Align iRMC driver with Ironic\u0027s default boot_mode\n\nThis commit modifies iRMC driver to use ironic.conf [deploy]\ndefault_boot_mode as default value of boot_mode.\nBefore this commit, iRMC driver assumes Legacy BIOS as default\nboot_mode and value of default_boot_mode doesn\u0027t have any effect\non iRMC driver\u0027s behavior.\n\nStory: 2010381\nTask: 46643\nChange-Id: Ic5a235785a1a2bb37fef38bd3a86f40125acb3d9\n(cherry picked from commit 071cf9b2dd2233509c3cdb8261c0ee7b391e6b2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/95341123a33f0bd58e56cff2fe494f39da01e947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/95341123a33f0bd58e56cff2fe494f39da01e947"}]},"branch":"refs/heads/bugfix/21.0"},"29546c18dccf19d80e2fac866677538d641f8be6":{"kind":"REWORK","_number":2,"created":"2022-12-07 22:45:17.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/56/866656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/56/866656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/56/866656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/56/866656/2"}}},"commit":{"parents":[{"commit":"fe64d5e1ecd3bdeab25232d4ab06ef984753547f","subject":"Merge \"Add support auth protocols for iRMC\" into bugfix/21.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fe64d5e1ecd3bdeab25232d4ab06ef984753547f"}]}],"author":{"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","date":"2022-10-25 05:14:44.000000000","tz":-240},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-12-07 22:45:07.000000000","tz":780},"subject":"Align iRMC driver with Ironic\u0027s default boot_mode","message":"Align iRMC driver with Ironic\u0027s default boot_mode\n\nThis commit modifies iRMC driver to use ironic.conf [deploy]\ndefault_boot_mode as default value of boot_mode.\nBefore this commit, iRMC driver assumes Legacy BIOS as default\nboot_mode and value of default_boot_mode doesn\u0027t have any effect\non iRMC driver\u0027s behavior.\n\nStory: 2010381\nTask: 46643\nChange-Id: Ic5a235785a1a2bb37fef38bd3a86f40125acb3d9\n(cherry picked from commit 071cf9b2dd2233509c3cdb8261c0ee7b391e6b2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/29546c18dccf19d80e2fac866677538d641f8be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/29546c18dccf19d80e2fac866677538d641f8be6"}]},"branch":"refs/heads/bugfix/21.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
