)]}'
{"id":"openstack%2Fironic~871997","triplet_id":"openstack%2Fironic~stable%2Fyoga~I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","project":"openstack/ironic","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-01-27 22:47:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","subject":"Move and fix reno config for releasenotes job","status":"MERGED","created":"2023-01-27 18:38:51.000000000","updated":"2023-01-27 22:48:21.000000000","submitted":"2023-01-27 22:47:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871997","cherry_pick_of_change":871996,"cherry_pick_of_patch_set":1,"meta_rev_id":"3a7cdde870fdb0ef7fd5e1e8bd53d4474e3a8624","_number":871997,"virtual_id_number":871997,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 22:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-01-27 19:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-01-27 19:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 19:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 19:39:31.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-01-27 20:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 21:46:28.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"}],"messages":[{"id":"a3cf94dd4d8c0d75e073b33df8e9801efffe1734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-27 18:38:51.000000000","message":"Patch Set 1: Cherry Picked from branch bugfix/20.2.","accounts_in_message":[],"_revision_number":1},{"id":"b0688152ad151e5db13660d3b151ce27e8655d55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:05:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672dede74bab48da83b9d5f760a5e807\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/371452a4d34541f090ee284558c659af : FAILURE in 18m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89dcee11190f4f41a41fee25cc8da481 : SUCCESS in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ead94614dcbf211659053baac1e9fa68398cb46e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-01-27 19:39:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45ec82d0dc08e1e99f264fa503a994227616085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:44:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1487b619f69044519c6f582c750fffa3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beb26ca9cd894a578790e98efa1ca33e : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8768bfb0574048dcaf78bfa0dd00eb57 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef87457fb9a4d09b5247b03ee45f2b7 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58b8e712edc34ac9be8f0807627e429e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c922eee8d16648d5834e57c15de62663 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f6a5e26c70e45109304f078f5792176 : SUCCESS in 5m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fa19e974e040483b9cc9e2377275f963 : SUCCESS in 11m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/99bd1808342b4cc6a6a6b3a8fc19cfab : SUCCESS in 7m 44s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/540bc4e5870e4f2b95966a28b2f76328 : SUCCESS in 30m 41s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/14cdb042ecda46a69bac2f88f19404c8 : SUCCESS in 25m 06s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/a45706abcca642039cc3a52359ba278e : SUCCESS in 1h 11m 42s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/ed22400198ea4f489c8c672234e90ea2 : SUCCESS in 46m 11s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f9864a09597d4ffeabad9504ff471035 : SUCCESS in 1h 06m 40s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0123ea73842e4fb1876028002f4b93bb : SUCCESS in 48m 59s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a44328917aac4e328f5e6298b8bc14b5 : SUCCESS in 47m 54s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/36341ce3aa604a6bacefcb3dea913130 : SUCCESS in 45m 23s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/5c62aff9ef244b59a84447c958bf7aaa : SUCCESS in 1h 15m 41s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/743192a9ba0343499a1e730479ab7896 : SUCCESS in 1h 08m 50s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d5c426ddc9234e38a6d5f268bb8989e4 : SUCCESS in 2h 02m 03s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0098c3b48677499c91448b408fa24318 : SUCCESS in 47m 58s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/8a17e97e9a1c46ecbf688327b507cdf5 : SUCCESS in 36m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/511daf60cf754737a4e23376e36f6569 : SUCCESS in 40m 53s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/919bf0f915624e89976e6a9a1cd6ef9b : SUCCESS in 1h 25m 17s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/f231027ffffc4c27a6be562a8dbd32e6 : SUCCESS in 4m 09s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/62bc4c734e3846309da82f0fe2a740b6 : SUCCESS in 1h 00m 30s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/6a51d8a77da842fdbdc7746f0a8d3067 : FAILURE in 1h 18m 54s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c567d8b718ae4f309127f1ce11cfcab3 : FAILURE in 54m 13s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f2aa1eec5bb449b3b33744394bb17487 : SUCCESS in 59m 43s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/c8133b094ec3456b8de8c48739f47837 : SUCCESS in 1h 01m 21s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c2e2cba87d44720a453115e455ff80b : SUCCESS in 25m 54s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-8 https://zuul.opendev.org/t/openstack/build/5c7dfc921d57487994c8a9eed6dde120 : SUCCESS in 31m 26s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/6f370e3c753f41698ffaa1ad30c3ec9e : FAILURE in 1h 27m 20s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/33b55f6a8ab8444daf2fc1f9c607d7cf : FAILURE in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a5c07b275efd67b37e2af4bb21c87c712a1af22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"545d1bb89be7647fd162391589ce81d32a2f577c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-27 21:46:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/871997/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/fa76570/fa76570.html : FAILURE in 8m 08s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/871997/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/0c99a0e/0c99a0e.html : FAILURE in 10m 19s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/871997/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/da0a63b/da0a63b.html : FAILURE in 13m 43s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/871997/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/ca55636/ca55636.html : FAILURE in 12m 46s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/871997/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/2c9dbbb/2c9dbbb.html : FAILURE in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03e5da5ec0b161d911c5602f17699fa0d2ab423a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:47:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0774c5b9788d461bb3d245d78a7cf1fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2970b8ffcc024d11a84bbdad9e96d539 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c70e14fe8a146e7bfc16821af5e08f1 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82bcd1c9fc4e4f63bb84621be01615a0 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511e5ea1d6fb4f90baeb148e657a56f7 : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2879eed4e3148f285a0535abb1b0e77 : SUCCESS in 6m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a7fa182c682d4f3297f31dae64f4eda9 : SUCCESS in 12m 11s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c07b53d6987748668de8b4f8f152d360 : SUCCESS in 19m 53s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4fec3818d0744471a1269f318867d0fc : SUCCESS in 26m 49s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/a99ce36f26e84ffd91563c889c4bf418 : SUCCESS in 1h 17m 58s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1ebfd476786548a492ea817114d3c515 : SUCCESS in 1h 04m 55s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/f312b2090855471a85a1a507a3265902 : SUCCESS in 1h 03m 24s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c58098dcc3ce471080f9a41d4349ad8a : SUCCESS in 43m 18s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/255cfed016e44a5b96ff052e73918a6f : SUCCESS in 57m 07s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/ad5d9706f4de4741b0880f62317526b4 : SUCCESS in 50m 17s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/30cce34537004553a1d3829fa6c48e74 : SUCCESS in 1h 56m 51s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/cf70ee37e00b483094cface0e8a0ef40 : SUCCESS in 46m 11s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/23c186f04a5f4f6daa79ceba4921c181 : SUCCESS in 44m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f48a45f58e1047a2ada483627b3e07fc : SUCCESS in 48m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/f857fb8b49004647bd82e34fe4ba2461 : SUCCESS in 56m 29s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/388848e0b9494dc59b665db19643d1f0 : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1bf1d6fcf20903e7e426bbe08efb97ceb80e3296","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:47:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a7cdde870fdb0ef7fd5e1e8bd53d4474e3a8624","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f973b027434cdba416609abbb6e2d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168a49235b594c989ffe73e2aec454eb : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58b13547e06248fe8f473f8d1dfa75a8 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5267b58e15d4f7b2fd2538ed6621256072b861e","revisions":{"e5267b58e15d4f7b2fd2538ed6621256072b861e":{"kind":"REWORK","_number":1,"created":"2023-01-27 18:38:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/97/871997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/97/871997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/97/871997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/97/871997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/97/871997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/97/871997/1"}}},"commit":{"parents":[{"commit":"26573bed3aa64f8fdb81cada528d30373060c766","subject":"Prevent pxe retry when agent token exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/26573bed3aa64f8fdb81cada528d30373060c766"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-26 23:36:14.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-27 18:38:51.000000000","tz":0},"subject":"Move and fix reno config for releasenotes job","message":"Move and fix reno config for releasenotes job\n\nIn its current place, reno config changes will not cause\nbuild-openstack-releasenotes job to run, which means changes can land to\nthat config without being tested. Yikes!\n\nAlso fixes error in regexp which was preventing this from actually\nfixing the build-openstack-releasenotes job.\n\nCo-Authored-By: Adam McArthur \u003cadam@mcaq.me\u003e\nChange-Id: I4d46ba06ada1afb5fd1c63db5850a1983e502a6c\n(cherry picked from commit fbe22b23bc3f1c89822e17417747378c9449ebeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e5267b58e15d4f7b2fd2538ed6621256072b861e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/e5267b58e15d4f7b2fd2538ed6621256072b861e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
