)]}'
{"id":"openstack%2Fironic~872001","triplet_id":"openstack%2Fironic~stable%2Fwallaby~I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","project":"openstack/ironic","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-01-27 22:35:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","subject":"Move and fix reno config for releasenotes job","status":"MERGED","created":"2023-01-27 18:42:43.000000000","updated":"2023-01-27 22:37:18.000000000","submitted":"2023-01-27 22:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872001","cherry_pick_of_change":872000,"cherry_pick_of_patch_set":1,"meta_rev_id":"72f361c098bb4133af002b6855e767275b38d59d","_number":872001,"virtual_id_number":872001,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 22:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-01-27 19:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-01-27 19:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 18:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 19:43:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-01-27 20:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 21:51:22.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"}],"messages":[{"id":"51ed1dd9754ea33623584fb44c1f2c870e0a7243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-27 18:42:43.000000000","message":"Patch Set 1: Cherry Picked from branch bugfix/18.1.","accounts_in_message":[],"_revision_number":1},{"id":"c5b1aa7f163b57e7f798fc5593bb49c45516daec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 18:51:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ea5619e03264ba88e1a3eb5aadeadcb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbab2abe117b4174b8424673fb2f33bc : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6eb02881fc7a4f28b90af3098edc5389 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdf8dba729aa60559ad612a392c31617db157d40","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-01-27 19:43:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31f091811a656bdcde9d262473834a22693cb4d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:36:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c555e850873340e29754b77d79c42762\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa39b0a10a14dd695cede514044c2f0 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655f6305471f401a98ad4148157181dd : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc021b120b548d183487ca4991c52cd : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/032823dfe22a4890a0783bdde947ea67 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ccfc34bccbd49f0b7594885bbd916d7 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f17bd3274fd4e5d83afdc570c49c59a : SUCCESS in 17m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d0265559f944b7935e40b03b2537a4 : SUCCESS in 5m 59s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/890ca1e4c40e42a38b96e47ea0987b04 : SUCCESS in 11m 11s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a2a0fceb3690438d9205b54f3da61dbe : SUCCESS in 28m 13s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/031b5a7bc9f242b0b26355fd6d8a5f63 : SUCCESS in 1h 22m 03s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/953a78cf409e462fa0dfee2ebae3cb61 : SUCCESS in 1h 08m 22s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/08b427beb70a4c09a51cf0a2d676d4fe : SUCCESS in 1h 29m 26s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/91d1db74b0c54ab298aa29d10af56020 : SUCCESS in 54m 04s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/72bdf30609f94a0998ec54695f1ce9ee : SUCCESS in 1h 05m 30s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/2cd4801b283c4ed1b9ab87bbd06035f5 : SUCCESS in 47m 24s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/8644f6bcd58044f696228819acbeb2be : SUCCESS in 1h 34m 49s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/5ac6f8e5db5a430eafff1d5d50ed02ca : SUCCESS in 54m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/c1b036a5636841809547bb001e6cdfd8 : SUCCESS in 1h 42m 24s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c52b00ece8cd4ae387728a6ee9e9fff1 : SUCCESS in 47m 34s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/acf39a75c786455dafe67ade2b9641f6 : SUCCESS in 51m 26s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c6fa655f409c4e009f3eadc5809469cf : SUCCESS in 46m 46s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/8c52f7a352a04911924bb027850ea8f1 : SUCCESS in 1h 47m 31s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/872025adf3cc450a8cd9288c24001442 : FAILURE in 4m 12s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/602743e501b24170be4bbbeda0235492 : SUCCESS in 40m 42s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/700fef9e3a1c4c3ea7da7ec6aa5eb098 : SUCCESS in 57m 52s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9dc61c1875504e53a0944ef0ddb1593b : FAILURE in 1h 02m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/d3cf5302958a49538de8b6b2a0e4c301 : SUCCESS in 49m 41s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8680cba78d44f6883d37358796bff30 : SUCCESS in 22m 57s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/6dca71ac07ae414ea8c4220fe929afcf : FAILURE in 1h 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b6fd470f7e3fefef7bebfea7a6d827682b15885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9035b8464bf3c743999a98cacd091ad564416d37","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-27 21:51:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/01/872001/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/1be9676/1be9676.html : FAILURE in 8m 12s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/01/872001/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/299b8da/299b8da.html : FAILURE in 16m 47s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/01/872001/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/39d8ec1/39d8ec1.html : FAILURE in 12m 33s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/01/872001/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/253be6e/253be6e.html : FAILURE in 7m 47s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/01/872001/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/76433bf/76433bf.html : FAILURE in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd796a03d383e790a1f07a85b91b6d93a7714c2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:35:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221a4a87f12a4ae4b8fe2b04728b0498\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637745a826054bf9a4d04c9c36b66348 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f459af75463c475b8e92607658be63d8 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce327e06001147418c0c18f7b86dd402 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c22e8fa0c74e499f89c717e4e064ac : SUCCESS in 15m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/574df352b2ab4467b9cc18b7e42d0847 : SUCCESS in 7m 10s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/524ab914fe144c1c8b2ba82fa2f844dc : SUCCESS in 10m 43s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b2da66e23091449784480a85cb65277f : SUCCESS in 24m 52s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/a806b318819c479f99ad2aac7977520f : SUCCESS in 1h 15m 08s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/6a439bd957f144fbbaa24c04afdc029c : SUCCESS in 1h 02m 37s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/485f8266ff6b44a4b5f375904ac7d795 : SUCCESS in 1h 22m 45s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3bc202868cb14cd091c1557e00790b6c : SUCCESS in 38m 03s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ef0c18b45a734acea846f652a7e4a652 : SUCCESS in 1h 02m 53s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/f6770ec6a61b4f129a6ebd50f17c730c : SUCCESS in 43m 31s\n- ironic-tempest-ipa-partition-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/9bb9a7598ea042389cefc98b7beb9955 : SUCCESS in 1h 31m 20s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool https://zuul.opendev.org/t/openstack/build/ea8e17edc0b744a2b016d9e6023da2ad : SUCCESS in 41m 24s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ee130d90df6a4d59891e693d26be8a1b : SUCCESS in 55m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/10006eb0543c4298b80d728e899919ca : SUCCESS in 30m 25s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/368838183903482cb836e969aae062be : SUCCESS in 50m 39s\n- metalsmith-integration-glance-localboot-centos8-uefi https://zuul.opendev.org/t/openstack/build/3870d34b736848858a828e0c20b0220a : SUCCESS in 1h 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e5afdfb020070f8e8d073f3bc1ebb904fc86e21c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:35:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"72f361c098bb4133af002b6855e767275b38d59d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 22:37:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25170214c7b4696bb5da9b98c5b8f52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e1b73da2b946d19596e9abe3bb7979 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93836f06c594477af4a0baf8c766536 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37b4b207f851f62176fafe1097a15e17b7b6648a","revisions":{"37b4b207f851f62176fafe1097a15e17b7b6648a":{"kind":"REWORK","_number":1,"created":"2023-01-27 18:42:43.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/01/872001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/01/872001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/872001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/872001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/01/872001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/01/872001/1"}}},"commit":{"parents":[{"commit":"06821f4d0cd1c217ecc44d2f41cd93ac05b5b236","subject":"Prevent pxe retry when agent token exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/06821f4d0cd1c217ecc44d2f41cd93ac05b5b236"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-26 23:36:14.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-27 18:42:43.000000000","tz":0},"subject":"Move and fix reno config for releasenotes job","message":"Move and fix reno config for releasenotes job\n\nIn its current place, reno config changes will not cause\nbuild-openstack-releasenotes job to run, which means changes can land to\nthat config without being tested. Yikes!\n\nAlso fixes error in regexp which was preventing this from actually\nfixing the build-openstack-releasenotes job.\n\nCo-Authored-By: Adam McArthur \u003cadam@mcaq.me\u003e\nChange-Id: I4d46ba06ada1afb5fd1c63db5850a1983e502a6c\n(cherry picked from commit fbe22b23bc3f1c89822e17417747378c9449ebeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/37b4b207f851f62176fafe1097a15e17b7b6648a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/37b4b207f851f62176fafe1097a15e17b7b6648a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
