)]}'
{"id":"openstack%2Fironic~872003","triplet_id":"openstack%2Fironic~stable%2Fussuri~I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","project":"openstack/ironic","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-01-27 21:40:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","subject":"Move and fix reno config for releasenotes job","status":"MERGED","created":"2023-01-27 18:43:10.000000000","updated":"2023-01-27 21:41:34.000000000","submitted":"2023-01-27 21:40:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872003","cherry_pick_of_change":872002,"cherry_pick_of_patch_set":1,"meta_rev_id":"9c82ea3b2dd622f51713e9e809eb64af258b12e4","_number":872003,"virtual_id_number":872003,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 21:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-01-27 19:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-01-27 19:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 19:43:48.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-01-27 20:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 21:10:14.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"}],"messages":[{"id":"109b3495769394a9a89a15a12f0638a87859073d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-27 18:43:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"807c838bf4f1fc23e571bb14ee228b11f2d31faa","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-01-27 19:43:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1bbd1aab369f015d64e228a0d7a7232019ec4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914113e708a14c5a832cbaefcdd00fe0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a7cf16e7401492faa1b9547d7b30d27 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb211517cc234fc981ec33e3e33a41f8 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1902a2184f2b4944b7b909a745ef5449 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2129d028832e44008c5d278808de6369 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897f0ff03efe4e8aa70cbd069de03c6d : SUCCESS in 19m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8cded194e24940b29cff0ab43c29c8 : SUCCESS in 6m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c1fa549fbf9141f1aa4e8cf8a4046adc : SUCCESS in 10m 03s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/f9e31764403449119c20fd61c8d4065a : SUCCESS in 1h 03m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ba727a75c9e349599d3f79fc8961f53f : SUCCESS in 38m 16s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7d6915d581784cfea00b44ac0b0a1298 : SUCCESS in 44m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/cbea547e7ef64a028d69636a838993c1 : SUCCESS in 47m 40s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/ac28c545e2ad4f2c85e258a63bd0df4d : SUCCESS in 1h 00m 06s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/64aae51c93644fafb2e7ddf7027ac3e4 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d9daa268940539f551ca51e70e39a425be42790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 20:04:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb6bc051cf47b5ebd1f38213504cad89b4372623","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-27 21:10:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/03/872003/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/58c50de/58c50de.html : FAILURE in 7m 32s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/03/872003/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/8ccaf8d/8ccaf8d.html : FAILURE in 13m 29s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/03/872003/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/edc859e/edc859e.html : FAILURE in 16m 07s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/03/872003/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/902d6a2/902d6a2.html : FAILURE in 14m 16s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/03/872003/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/3a6e9d5/3a6e9d5.html : FAILURE in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d114b4d1b3fb5fd9c6905c0ee7bdb165934cb4d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 21:40:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb93fead4b3e41ebbd6cc5d978fd838d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca26ae32ede41578aa7d6b5281fb59d : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b26d0ed3e7d14bbc9ae5bd4bc9376638 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4547dae9367400f8ed69d88ea75850a : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c75691432a4509b9bbf8b2295ca99e : SUCCESS in 14m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34f405143bab4504a326648d2629f678 : SUCCESS in 5m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d894ed302d1045e19bfcd3d0faf1719d : SUCCESS in 8m 34s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/7804ce594a0d42e3ac580b731e9c7518 : SUCCESS in 1h 07m 01s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cbddfb1a54064537a49bb70628881a7d : SUCCESS in 23m 57s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5dbf0697c65f4be8a177c4092665d72c : SUCCESS in 1h 11m 01s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/da17ff9ca7da43aeb238fde8ac760330 : SUCCESS in 33m 07s\n- metalsmith-integration-glance-localboot-centos7 https://zuul.opendev.org/t/openstack/build/94a828c966dc4c03829c0552bb39c0df : SUCCESS in 1h 30m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3e958ba8967272d21103e006cddcbfb52699a0e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 21:40:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9c82ea3b2dd622f51713e9e809eb64af258b12e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 21:41:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/906236b3a38248dd9209bdb9aaf0919d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d74d04e19dc415e8849f2c32cb0624b : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/074e7cb8ed3440cd960bc84d7a4e8b86 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"859e51c8b4b8344827b5bba1f9a0b737ffbc1ebc","revisions":{"859e51c8b4b8344827b5bba1f9a0b737ffbc1ebc":{"kind":"REWORK","_number":1,"created":"2023-01-27 18:43:10.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/03/872003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/03/872003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/872003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/872003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/872003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/03/872003/1"}}},"commit":{"parents":[{"commit":"2571e3cf4ec804e31ff5afa11883034a700f8704","subject":"Fixes for tox 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2571e3cf4ec804e31ff5afa11883034a700f8704"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-26 23:36:14.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-27 18:43:10.000000000","tz":0},"subject":"Move and fix reno config for releasenotes job","message":"Move and fix reno config for releasenotes job\n\nIn its current place, reno config changes will not cause\nbuild-openstack-releasenotes job to run, which means changes can land to\nthat config without being tested. Yikes!\n\nAlso fixes error in regexp which was preventing this from actually\nfixing the build-openstack-releasenotes job.\n\nCo-Authored-By: Adam McArthur \u003cadam@mcaq.me\u003e\nChange-Id: I4d46ba06ada1afb5fd1c63db5850a1983e502a6c\n(cherry picked from commit fbe22b23bc3f1c89822e17417747378c9449ebeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/859e51c8b4b8344827b5bba1f9a0b737ffbc1ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/859e51c8b4b8344827b5bba1f9a0b737ffbc1ebc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
