)]}'
{"id":"openstack%2Fironic~872004","triplet_id":"openstack%2Fironic~stable%2Ftrain~I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","project":"openstack/ironic","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-04-27 15:13:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4d46ba06ada1afb5fd1c63db5850a1983e502a6c","subject":"Move and fix reno config for releasenotes job","status":"ABANDONED","created":"2023-01-27 18:43:20.000000000","updated":"2023-04-27 15:13:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":872003,"cherry_pick_of_patch_set":1,"meta_rev_id":"60c232f84987795e9ef6c77ba5714b30b3ac1cad","_number":872004,"virtual_id_number":872004,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-27 19:41:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-20 14:06:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 19:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 21:24:52.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-02-20 14:06:56.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"75eeb594e027c9d871d096f446260540eb5aa539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-27 18:43:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"ff368f6128df0a33798febc528929829efbef64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:41:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/780ea067980c4498a9c8d00002f8e5db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4cb88e230a24302be39968890566eff : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09df3f04591d43a3b3db6c1190a79f94 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93313e3c4041413a85034a7bcdd765f6 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39617b7b13c04708afb6434951a2fb16 : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63679a6bea7545ca80d0ca498066a3b5 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4e5d487f434e2286097a2a506cb576 : SUCCESS in 16m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b12c5ddd050c4066881b99391937ccc4 : SUCCESS in 6m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c53b2abf111e47f1ae41edf7e5339301 : SUCCESS in 10m 35s\n- ironic-tox-unit-with-driver-libs-python3 https://zuul.opendev.org/t/openstack/build/e9db121898544783a964e800cf8c8f92 : SUCCESS in 9m 24s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/527666a7b6ca4172b302bc8c2f319af3 : FAILURE in 31m 45s\n- ironic-tempest-functional-python2 https://zuul.opendev.org/t/openstack/build/3b90b27ba59a44738f82bd83ef55466d : FAILURE in 24m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/3c274e84c2054452a4fc2f171f541bc5 : FAILURE in 26m 43s\n- ironic-tempest-ipa-partition-pxe_ipmitool-tinyipa-python3 https://zuul.opendev.org/t/openstack/build/24cc0af6fcc149dc89b26f2cbaeba45a : FAILURE in 18m 57s\n- ironic-tempest-ipa-partition-redfish-tinyipa https://zuul.opendev.org/t/openstack/build/d7df80437d694f7eb71419779283df41 : FAILURE in 32m 47s\n- ironic-tempest-ipa-partition-uefi-pxe_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/56c20ecd4fd14cd6b4e68073a77d6eaa : FAILURE in 20m 00s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/365da27ae715467b829455b7a07b6eae : FAILURE in 29m 03s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2ffbfec9fa5c4d22bfb7564132e5c329 : FAILURE in 20m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/6c43d9b24ff44ae382beaf2e283c5547 : FAILURE in 20m 14s\n- ironic-tempest-ipa-partition-bios-agent_ipmitool-tinyipa-indirect https://zuul.opendev.org/t/openstack/build/502d6f1558b74cac955e63b0c689d475 : FAILURE in 19m 03s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/ce8dcf7e02264d6d81b98a9bcb7d28b1 : FAILURE in 23m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/8db3dec192e94c0f94fbfb05f7f34393 : FAILURE in 19m 11s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/a2fc353010eb47de9eda6ba36c2ba477 : SUCCESS in 4m 30s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-pxe_snmp-tinyipa https://zuul.opendev.org/t/openstack/build/5951009e1bc84216b3c95005342a2538 : FAILURE in 30m 04s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b212a702b0b9435c9b73da3ef9704a77 : FAILURE in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c11d6b1a8d1460a8769c597f40275cf974718b23","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-27 21:24:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/04/872004/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-virtual-media-direct-python3/30261bb/30261bb.html : FAILURE in 15m 21s (non-voting)\n- hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/04/872004/1/check/hpe-gen9-dsvm-ironic-standalone-wholedisk-uefi-ilo-pxe-direct-python3/1d4cd7d/1d4cd7d.html : FAILURE in 8m 02s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/04/872004/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/f985f25/f985f25.html : FAILURE in 13m 45s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/04/872004/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-uefi-redfish-pxe-python3/133c2c7/133c2c7.html : FAILURE in 12m 20s\n- hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/04/872004/1/check/hpe-gen10-dsvm-ironic-standalone-wholedisk-ilo-uefi-https-direct-python3/30f9924/30f9924.html : FAILURE in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae31ac9e32df0bf38e18d00eb45768ec0f613706","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-02-20 14:06:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60c232f84987795e9ef6c77ba5714b30b3ac1cad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-04-27 15:13:06.000000000","message":"Abandoned\n\nThis is not great, but I\u0027m not going to fix CI for train to fix this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"945fc40b31a64776c713b884eb3e6706371aa39a","revisions":{"945fc40b31a64776c713b884eb3e6706371aa39a":{"kind":"REWORK","_number":1,"created":"2023-01-27 18:43:20.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/04/872004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/872004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/872004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/872004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/872004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/872004/1"}}},"commit":{"parents":[{"commit":"4d1fa79e18a3e9be1aa9e23c21e8874788a4013b","subject":"Fixes for tox 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4d1fa79e18a3e9be1aa9e23c21e8874788a4013b"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-26 23:36:14.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-01-27 18:43:20.000000000","tz":0},"subject":"Move and fix reno config for releasenotes job","message":"Move and fix reno config for releasenotes job\n\nIn its current place, reno config changes will not cause\nbuild-openstack-releasenotes job to run, which means changes can land to\nthat config without being tested. Yikes!\n\nAlso fixes error in regexp which was preventing this from actually\nfixing the build-openstack-releasenotes job.\n\nCo-Authored-By: Adam McArthur \u003cadam@mcaq.me\u003e\nChange-Id: I4d46ba06ada1afb5fd1c63db5850a1983e502a6c\n(cherry picked from commit fbe22b23bc3f1c89822e17417747378c9449ebeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/945fc40b31a64776c713b884eb3e6706371aa39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/945fc40b31a64776c713b884eb3e6706371aa39a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
