)]}'
{"id":"openstack%2Fironic~877403","triplet_id":"openstack%2Fironic~bugfix%2F21.0~I236cdf6ddb040284e9fd1fa10136ad17ef665638","project":"openstack/ironic","branch":"bugfix/21.0","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-07-31 15:00:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I236cdf6ddb040284e9fd1fa10136ad17ef665638","subject":"Wipe Agent Token when cleaning timeout occcurs","status":"ABANDONED","created":"2023-03-15 08:31:30.000000000","updated":"2023-07-31 15:00:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":877395,"cherry_pick_of_patch_set":1,"meta_rev_id":"0b6d6dbd68756f957719cb1897c3a5afde727418","_number":877403,"virtual_id_number":877403,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-04-12 10:51:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":2,"date":"2023-03-16 22:40:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-12 10:36:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-12 10:36:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 08:31:30.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-03-15 08:39:17.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-03-15 08:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 22:40:28.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-04-12 10:36:22.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"167197bea12d50c1dfe3d8c9abf488e743cedbd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-15 08:31:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"e6c9e3b243a92f79a36c95f1c864690e6bc4b587","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-03-15 08:39:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/03/877403/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/56a0521/56a0521.html : FAILURE in 5m 28s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/03/877403/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/ab3c215/ab3c215.html : FAILURE in 6m 23s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/03/877403/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/9d89479/9d89479.html : FAILURE in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be934b77a8da588e36fcfae833cd23f911234182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 08:55:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3afd218b3654284bbcd631ee0577332\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6cb2fe8b56e429b940d6a40d59d33a7 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94617222184e45d3b88fc42946f7680d : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e04ccb2f66874ee7a92db93eaf0c3f01 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8326b16f46ae4a90a3a89d7584dbd5dc : SUCCESS in 8m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fa237fba5934baf9f95a590ae201951 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d807e4af6f424ad29b167646501b5960 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b06188d6198437085107791d27023c1 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4820e33b3d42407abbc8434c2784618b : SUCCESS in 13m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ef7228714b6f456bb393aced08aeaaa1 : SUCCESS in 10m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/5243835b83ae4ea49e340743e9432077 : FAILURE in 3m 08s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/d3f9a3806527441e8232101cead26a92 : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7121a8bf38f71e0fe9f30822ab199de80daea256","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-03-16 22:40:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df56894ca36403edb7490c0da257c1c88f5581bd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-04-12 10:36:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d60d7b4c0e1fb339f4a493d0c040790e46abace2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 10:36:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d61bbc29736504732e5b9c7f920415f30a63b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 10:51:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98a19beda9464a1d989a4ed437df63cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13b00807ad6c489abcd98816f7dec80f : FAILURE in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777353c579cb45c391d34b8fa94c3622 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/824590758d924dccb82f1b16d0aa70f9 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ad2f75916142218fb551a2057194e0 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297047cfdf1646b983b09d7661a4de45 : SUCCESS in 12m 38s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/fb35d9844fab4c788595b9ffe12e8a73 : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0b6d6dbd68756f957719cb1897c3a5afde727418","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-07-31 15:00:49.000000000","message":"Abandoned\n\nbranch not supported anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e460f9c67ce55c38733405690d7135448f798c0","revisions":{"9e460f9c67ce55c38733405690d7135448f798c0":{"kind":"REWORK","_number":1,"created":"2023-03-15 08:31:30.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/03/877403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/03/877403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/877403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/877403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/03/877403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/03/877403/1"}}},"commit":{"parents":[{"commit":"7ea86a8669ddcba70b9a3ca05f54ebd341446097","subject":"Do not move nodes to CLEAN FAILED with empty last_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7ea86a8669ddcba70b9a3ca05f54ebd341446097"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-02-28 21:28:49.000000000","tz":-480},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-15 08:31:30.000000000","tz":0},"subject":"Wipe Agent Token when cleaning timeout occcurs","message":"Wipe Agent Token when cleaning timeout occcurs\n\nIn a relatively odd turn of events, should cleaning\nhave started, but then timed out due to lost communications\nor a hard failure of the machine, an agent token could\npreviously be orphaned preventing re-cleaning.\n\nWe now explicitly remove the token in this case.\n\nChange-Id: I236cdf6ddb040284e9fd1fa10136ad17ef665638\n(cherry picked from commit 47b5909486c336352c536eb2cadd121afea8cf12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9e460f9c67ce55c38733405690d7135448f798c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9e460f9c67ce55c38733405690d7135448f798c0"}]},"branch":"refs/heads/bugfix/21.0"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
