)]}'
{"id":"openstack%2Fironic~877404","triplet_id":"openstack%2Fironic~bugfix%2F20.2~I236cdf6ddb040284e9fd1fa10136ad17ef665638","project":"openstack/ironic","branch":"bugfix/20.2","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-04-12 12:32:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I236cdf6ddb040284e9fd1fa10136ad17ef665638","subject":"Wipe Agent Token when cleaning timeout occcurs","status":"MERGED","created":"2023-03-15 08:31:51.000000000","updated":"2023-04-12 12:33:57.000000000","submitted":"2023-04-12 12:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877404","cherry_pick_of_change":877395,"cherry_pick_of_patch_set":1,"meta_rev_id":"4efcc35a8d95babb1d136d753cb893715ccc690c","_number":877404,"virtual_id_number":877404,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 12:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-03-16 22:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-12 10:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-12 10:36:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 08:31:51.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-03-15 09:07:17.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-03-15 10:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 22:40:36.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-04-12 10:36:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"6a1b648d851dadb7901fb39d1547ed6634f72012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-15 08:31:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"2e5549a7dce747c60da58b363fc039f86a1ab1ea","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-03-15 09:07:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct finger://localhost.localdomain/8f98edaba82744a1b2d14664e5462fa6 : RETRY_LIMIT in 5s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/04/877404/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/2148f43/2148f43.html : FAILURE in 25m 18s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/04/877404/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/eec99e2/eec99e2.html : FAILURE in 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89545b125895894cc04f4d2c158c0c6f23e6208b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 10:54:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba62fb103fc54ad3ba164c8135f18656\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0be9ed3be0f4489b1dfeb9095c545e1 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11855f594fcd4c068ebcffcf788139db : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e42649035c423197183f7a124dca9e : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f914b6ab16a45809822e4acd69523d8 : SUCCESS in 14m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74fab97d95c44fccbb477e7b2510e835 : RETRY_LIMIT in 2m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7749bb6b8d3844a094dfad62b20aec6d : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/10788c4ec4974693bb20c1cf3280fba1 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69445cf2917040a580fbda3054ece3cc : SUCCESS in 13m 09s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/031da71a0e874910bd24ee6b2338d140 : SUCCESS in 10m 17s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/03d945a7be2e4ad99d161557be6f940e : RETRY_LIMIT in 2m 19s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/dc9b8df4c6fe46ca9cbc1863b3091796 : SUCCESS in 18m 21s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e956478ce4474bb587546f00a28ef292 : SUCCESS in 35m 00s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/00945553679846f187755cf88cce2483 : SUCCESS in 41m 53s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/980efea28e224eadafe928ba63294689 : SUCCESS in 50m 18s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b5df39d76b894ec09e4bc7f01947cd19 : SUCCESS in 57m 19s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/61a0578f0c1e42e89c8090e9021362e7 : SUCCESS in 51m 41s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e6ffe33e607d4faca10d1bcfe9035ee9 : SUCCESS in 44m 15s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3f38d13daa68439cb94fc12e39dd6a37 : SUCCESS in 1h 18m 34s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4373537edf784fcca7153c2219ddd929 : SUCCESS in 1h 05m 49s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/9e8f3b060215487d9e1850803aef2cbc : SUCCESS in 2h 12m 26s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/668b412c0ed346ab90d3e43d8248eb69 : SUCCESS in 1h 02m 43s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2c5b5314ef6a477d81a5c43e47e0ef25 : SUCCESS in 46m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a835fa6542254a6eb7ef034693d65116 : SUCCESS in 54m 28s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5dec339ffaf54d90ad5046a8d9efa1b7 : RETRY_LIMIT in 1m 56s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d562d4c578e7430dbc45c0fc4782ca8c : FAILURE in 1h 19m 11s (non-voting)\n- ironic-inspector-tempest-partition-bios-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2ed6c095d10c4c6d94593c6923378c86 : FAILURE in 46m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/a34386099ca04df2aceb7a9174285840 : SUCCESS in 43m 33s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e20081e34a28409ba12c5f939325a301 : FAILURE in 1h 04m 15s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/01a121a46c8644e8b49e088a73ca3df4 : FAILURE in 1h 02m 48s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/3f41a189c70c4fc3a9eb499e03d4583b : FAILURE in 1h 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49b68ce9195ba039d2278669d0a5b959704121a0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-03-16 22:40:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb08f7a5aa50446ada206655e221cb7618276138","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-04-12 10:36:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cc74e26d7150d7b6e7626011912e971fc9f8207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 10:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95211cbc958d11307a09997090d44be472f8c3ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 12:32:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9d9e23b76b4189beec6c32e50f0e66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3e9273902694dbc989aafdd34b9d37a : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3e380ad38e441297eb6777a2fe7f33 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d12e84be4049493fb13f4cbfd7678605 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cc20a107d0049a49c8e1080cd4cba54 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec14cbfa1af545be89e1845fd7a27dce : SUCCESS in 11m 53s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6a0b44e213784630afe299313ea22a03 : SUCCESS in 9m 46s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1df22079440148ac8a208e3e3d89ae29 : SUCCESS in 32m 05s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1789345837114f4194603fcdb41e6bcc : SUCCESS in 34m 30s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d454faebfd474f5bbf0dcb2f05e7558e : SUCCESS in 39m 20s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/87003ffe586d40e492e479091aca4504 : SUCCESS in 1h 23m 47s\n- ironic-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a2ed3a4d7c124cc090fa1031be735ea0 : SUCCESS in 1h 07m 35s\n- ironic-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c8a0103d9d5946a89ddf4209853e60d4 : SUCCESS in 51m 08s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3780dd935378486b9bb343c4eb339049 : SUCCESS in 52m 20s\n- ironic-tempest-partition-bios-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a2dcb0d638344054b0637c79f83a4a21 : SUCCESS in 1h 35m 09s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/bf5712a3c72044e9afefe08f8d97dba6 : SUCCESS in 56m 08s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/19183f5b13b1405dbd0f34418edfc7fb : SUCCESS in 52m 05s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/393b0dedba4f4027a1d875b4951c8ea9 : SUCCESS in 30m 21s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/a29001738c09420baae035bfb67020c0 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"64f21cb0a124a3cd31a38ec7781eb925634883d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 12:32:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4efcc35a8d95babb1d136d753cb893715ccc690c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 12:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2031b8dc0c84ad0b41a04dab52e4be3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd879f4bba35448b8e4240919feed085 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86d8e6e8892dfaa23c7f4a12b55c31032e9c6d77","revisions":{"86d8e6e8892dfaa23c7f4a12b55c31032e9c6d77":{"kind":"REWORK","_number":1,"created":"2023-03-15 08:31:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/04/877404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/04/877404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/877404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/877404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/04/877404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/04/877404/1"}}},"commit":{"parents":[{"commit":"280a1c9af1c05951c20db19c3f114e812beac75d","subject":"Do not move nodes to CLEAN FAILED with empty last_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/280a1c9af1c05951c20db19c3f114e812beac75d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-02-28 21:28:49.000000000","tz":-480},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-15 08:31:51.000000000","tz":0},"subject":"Wipe Agent Token when cleaning timeout occcurs","message":"Wipe Agent Token when cleaning timeout occcurs\n\nIn a relatively odd turn of events, should cleaning\nhave started, but then timed out due to lost communications\nor a hard failure of the machine, an agent token could\npreviously be orphaned preventing re-cleaning.\n\nWe now explicitly remove the token in this case.\n\nChange-Id: I236cdf6ddb040284e9fd1fa10136ad17ef665638\n(cherry picked from commit 47b5909486c336352c536eb2cadd121afea8cf12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/86d8e6e8892dfaa23c7f4a12b55c31032e9c6d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/86d8e6e8892dfaa23c7f4a12b55c31032e9c6d77"}]},"branch":"refs/heads/bugfix/20.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
