)]}'
{"id":"openstack%2Fironic~877407","triplet_id":"openstack%2Fironic~bugfix%2F21.0~Ie89e8be9ad2887467f277772445d4bef79fa5ea1","project":"openstack/ironic","branch":"bugfix/21.0","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-07-31 15:00:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie89e8be9ad2887467f277772445d4bef79fa5ea1","subject":"Clean out agent token even if power is already off","status":"ABANDONED","created":"2023-03-15 08:34:13.000000000","updated":"2023-07-31 15:00:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":877392,"cherry_pick_of_patch_set":1,"meta_rev_id":"b85dda94650a66498bdeee0e20827ca6250dc466","_number":877407,"virtual_id_number":877407,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-15 08:58:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 08:34:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-03-15 08:56:13.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-03-15 08:58:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec202584ea693ca663c83fcad28d7fbe4b74b335","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-15 08:34:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"33afd1d3776f102799885c780e8bc333b74b40b8","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-03-15 08:56:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct finger://localhost.localdomain/bd739ff6694745e1a59c102c16f19621 : RETRY_LIMIT in 5s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct finger://localhost.localdomain/81b3b6f21f1d4209861f5758dfdd55aa : RETRY_LIMIT in 21s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct finger://localhost.localdomain/ce3f2052230c4dcbae23ec4d965fbd29 : RETRY_LIMIT in 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e704a3ba7262ae1770de11eb39eacdc16eb9bd7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 08:58:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a7e63aa436498580e46facfbc26e43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d09f46d6ce843abb18fad739cff6bf9 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017c816a6c474e19bbe17d4141682346 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93896bb4de8e4c6a864017c4901b9702 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/612c609990634966b01db6c34a7ab2f3 : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7b38af8057844e0a14d4c88b61d5770 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0315767e35224251a2d5a34632c03cc9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96f42bb9916a4dfea09bc6a3e4562f3f : SUCCESS in 20m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d5a602d0f94d529e8e087ded48550d : SUCCESS in 19m 10s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/33a0184776e447739726840d5bf60c62 : SUCCESS in 9m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/62620c0d5bbb40a8af66cecae3c6e425 : FAILURE in 2m 49s (non-voting)\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/21593c3140ad4f3ea71a7af6806c9a8b : FAILURE in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b85dda94650a66498bdeee0e20827ca6250dc466","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-07-31 15:00:23.000000000","message":"Abandoned\n\nbranch not supported anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d40f6067de2f4c726c5f73b764ef2bacd1692b2c","revisions":{"d40f6067de2f4c726c5f73b764ef2bacd1692b2c":{"kind":"REWORK","_number":1,"created":"2023-03-15 08:34:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/07/877407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/07/877407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/877407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/877407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/07/877407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/07/877407/1"}}},"commit":{"parents":[{"commit":"7ea86a8669ddcba70b9a3ca05f54ebd341446097","subject":"Do not move nodes to CLEAN FAILED with empty last_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7ea86a8669ddcba70b9a3ca05f54ebd341446097"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-02-28 18:18:16.000000000","tz":-480},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-15 08:34:13.000000000","tz":0},"subject":"Clean out agent token even if power is already off","message":"Clean out agent token even if power is already off\n\nWhile investigating a very curious report, I discovered that\nif somehow the power was *already* turned off to a node, say\nthrough an incorrect BMC *or* human action, and Ironic were\nto pick it up (as it does by default, because it checks before\napplying the power state, then it would not wipe the token\ninformation, preventing the agent from connecting on the next\naction/attempt/operation.\n\nWe now remove the token on all calls to conductor\nutilities node_power_action method when appropriate, even\nif no other work is required.\n\nChange-Id: Ie89e8be9ad2887467f277772445d4bef79fa5ea1\n(cherry picked from commit bcf6c12269168c5b4f0d9d4d3212e813f1827494)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d40f6067de2f4c726c5f73b764ef2bacd1692b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/d40f6067de2f4c726c5f73b764ef2bacd1692b2c"}]},"branch":"refs/heads/bugfix/21.0"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
