)]}'
{"id":"openstack%2Fironic~877409","triplet_id":"openstack%2Fironic~stable%2Fzed~I7bec6cfacb9d1558bc514c07386583436759f4df","project":"openstack/ironic","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-16 12:03:24.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-03-16 12:03:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bec6cfacb9d1558bc514c07386583436759f4df","subject":"Fix online upgrades for Bios/Traits","status":"MERGED","created":"2023-03-15 14:13:03.000000000","updated":"2023-03-16 12:04:33.000000000","submitted":"2023-03-16 12:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877409","cherry_pick_of_change":876626,"cherry_pick_of_patch_set":4,"meta_rev_id":"b6f091c16edef025437102d1c99df9dcbfb0badd","_number":877409,"virtual_id_number":877409,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 12:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-16 09:16:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-16 10:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-16 10:30:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 14:13:03.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-03-15 14:20:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 14:48:26.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-03-15 17:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 09:16:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-16 10:30:40.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"08150fd147c581f2c24ba515feda25e96483b536","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-03-15 14:13:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ironic/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"189f768e1a5c2a216c77032ad4255374cb9bf714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:20:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/804535af9b414392a298b382534d4d39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/435588143ba141c8970fe6ea1077e0b7 : FAILURE in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a5e289a053e403dbbf4a9db9652b6f3 : FAILURE in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0317ae7c37ed4cc317b3b291f0b9aa27abc3fb3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-15 14:21:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6699ae50216c8bf3dc5af165e31592a50504ffc","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-15 14:22:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"6b35f279783664bb823fde5aa79840f6e257ea3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 14:33:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a26498fe6284b86b11813ec756d434d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24728f5ed00449abb1edb276baac0280 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a3eb6aef5c44b2ba9cda0258a97f295 : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"991197c3ec6b98c1ddfb9461cbfe50db71ee7525","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-03-15 14:48:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/646005f/646005f.html : FAILURE in 18m 33s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/0983ddd/0983ddd.html : FAILURE in 18m 17s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/3b0460c/3b0460c.html : FAILURE in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2978ccf7fab12b8faeb629202d0a9eaa496e86bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-15 15:41:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"73264340cd5858366644d8d612c576e7df092a4b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-03-15 15:57:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/3/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/f329188/f329188.html : FAILURE in 5m 21s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/3/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/1ecb65a/1ecb65a.html : FAILURE in 5m 38s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/09/877409/3/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/3c78173/3c78173.html : FAILURE in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"758f5a8517561503ba07f2126f89ad828a8193fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 16:07:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308c11bc02a24a2183a97478c8604a38\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56cdd7a0454444ad9fecde504efdc2d6 : SUCCESS in 23m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a644be971ce54e7a97faf17afe9b196c : SUCCESS in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"addad9e4e44b08b0c9cb268768d057c78eb0fa9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 17:21:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9076fd4ae854441b2c37fa2ff892f46\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2747466883da46c0851bd86ca2bc85b7 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9d0c1baae64a7fbaf872efff0c7e08 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b359093a5b140c2bb3b88b5322d77fb : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5200c66044a475b953c981987cc8d56 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea1c435a25954b84aa2caf39f1ea95ea : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0763788e989a4fb28ec0e21de24ee461 : SUCCESS in 15m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65ac77848e0145d1a53f56a485fdb16e : SUCCESS in 9m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/a90930e12eb44b92a097eb30c968f4f5 : SUCCESS in 11m 50s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d1eea8a8eec94ede90c2365d5b2b4d70 : SUCCESS in 9m 29s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cfed04a199bb4b9abbeda6ba5609fede : SUCCESS in 21m 03s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/dfbc45ef787a41c38d01a10a2a6361a7 : SUCCESS in 24m 57s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/18b3044f64a34b6bab250fae3afb48c4 : SUCCESS in 1h 00m 02s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/691afcd5775f40a6a6c434e54022cd50 : SUCCESS in 47m 13s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e14fee7536d34916972c5949e29667c2 : SUCCESS in 1h 03m 43s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fd576d14cce5413394717638ae569b36 : SUCCESS in 37m 21s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b2f432307bc54cd8b5a37341e1bb1039 : SUCCESS in 1h 01m 05s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/9a12c376309b494d81d218ec2f866465 : SUCCESS in 45m 19s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a9d31e25495d4de894ddaff38ae29459 : SUCCESS in 45m 19s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1fedec182c36486da542a4442e099c77 : SUCCESS in 1h 35m 21s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/a0ff53c290164baf86e4b41cae9bdc2a : SUCCESS in 1h 06m 25s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/3867381878cb484f8fcb65d9162d100d : SUCCESS in 42m 23s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/08f1ec279ac940d78954b890e4284af0 : SUCCESS in 54m 47s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/3a1cadb3db7b42589d85d195fdbb2ffc : SUCCESS in 1h 25m 26s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/54b043eb8fdc4cac990d0f47acf71fe4 : FAILURE in 4m 19s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0c45fa2d5a1f41f8a3ab6ed624d8ed88 : SUCCESS in 59m 25s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/f5980d31d94949fdb1507d16a71350a1 : FAILURE in 43m 35s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/2c9a1626062f401ebef28c68fb818d8d : SUCCESS in 51m 18s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/a71ea206d7e5480aab6a96afec564a03 : FAILURE in 1h 01m 54s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/172d49a5a38c41f6aa52f76737285c48 : FAILURE in 1h 04m 02s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/15c7d089fda14d869919ab4ea3426093 : SUCCESS in 54m 59s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2984d4b46f2d4dd9bf89100e44f824f5 : FAILURE in 11m 13s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4838ed80b1c64e41ba7155e10b0c790e : SUCCESS in 35m 43s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/00640d52e4b44676a0ce3b53a14a33d1 : SUCCESS in 41m 40s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/54315488eca04633bcf54b2608dbb3bb : FAILURE in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27b81f284b82a45ab2bcc23eb8a41b798a80aad7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-16 09:16:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"52bdabaf2109115c30370a2bb8777d46a6350c99","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-16 10:30:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53fb3b7a229ac97cf32a65f176f638cbbdba579e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:31:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51e510eba0cbe6146db45b30a0e4087b76268535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 12:03:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c108c0f5badc4882a57a56f36108008b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12308900656e49889245dae0230a09eb : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0563c0ee57f4764a3bec7453c8de1af : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f6310ff088247cf9ce9f8ae266fdac0 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc0397ce6314e6cac6b3807c50b93f6 : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbd532cd8ebe40249493247c942e1056 : SUCCESS in 7m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/299da55e37244074805ae5a4931f615d : SUCCESS in 9m 32s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c6adfa21a2144450acd6126d00d4cb1b : SUCCESS in 29m 04s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/74d5178c320f44baaeba52d5bb442876 : SUCCESS in 28m 52s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/4b9ba3ae22d2401abf379767371560b6 : SUCCESS in 51m 39s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/979b19e0705145ba9c50751b9cccd86b : SUCCESS in 55m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/2a671eaedc2e4e25a429804320a9bb9a : SUCCESS in 1h 01m 18s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ed734959ee1841768a4e529b15686f85 : SUCCESS in 39m 17s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bb15ab54c42143e49441a6cd3d72f273 : SUCCESS in 49m 38s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/42583b75e71c4e5abcb710baa268bcf0 : SUCCESS in 38m 12s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/2e2a889e663249a4b62a8971a81ad2ce : SUCCESS in 55m 12s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/a49807615f934a32b55d903c084ffa28 : SUCCESS in 59m 05s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/85f4b3769ff84275be9d76b12beb683d : SUCCESS in 36m 11s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/831cf78fd3214e54906a0d9c5c30ed7e : SUCCESS in 57m 53s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/7169ca8500fb4382bfc57e96817921e9 : SUCCESS in 1h 25m 46s","accounts_in_message":[],"_revision_number":3},{"id":"7f502d291e365329eb6c9369652741bee6dc6d72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 12:03:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b6f091c16edef025437102d1c99df9dcbfb0badd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 12:04:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc0b93526c2444cb31c8c0d59af3113\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18d8daa0b6542b7845bcadea8b5358f : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b69c85986f471fb1db23cd3c2c7b3b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b0a5575ce30e373f709388c2697fd9775b2850b","revisions":{"f3b13981ee6bff822d9e52875d3a8a216248d70b":{"kind":"REWORK","_number":1,"created":"2023-03-15 14:13:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/09/877409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/09/877409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/09/877409/1"}}},"commit":{"parents":[{"commit":"9802342616d3954832f14780ec2b74e0e59de932","subject":"Merge \"Wipe Agent Token when cleaning timeout occcurs\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9802342616d3954832f14780ec2b74e0e59de932"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-03-06 18:48:07.000000000","tz":-480},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-03-15 14:13:03.000000000","tz":0},"subject":"Fix online upgrades for Bios/Traits","message":"Fix online upgrades for Bios/Traits\n\n... And tags, but nobody uses tags since it is not available\nvia the API.\n\nAnyhow, the online upgrade code was written under the assumption\nthat *all* tables had an \"id\" column. This is not always true\nin the ironic data model for tables which started as pure extensions\nof the Nodes table, and fails in particular when:\n\n1) A database row has data stored in an ealier version of the object\n2) That same object gets a version upgrade.\n\nIn the case which discovered this, BIOSSetting was added at version\n1.0, and later updated to include additional fields which incremented\nthe version to 1.1. When the upgrade went to evaluate and iterate\nthrough the fields, the command failed because the table was designed\naround \"node_id\" instead of \"id\".\n\nStory: 2010632\nTask: 47590\n\nChange-Id: I7bec6cfacb9d1558bc514c07386583436759f4df\n(cherry picked from commit cbe5f86ce7c100f834701cf43f7b9b1bfdfb9ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f3b13981ee6bff822d9e52875d3a8a216248d70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f3b13981ee6bff822d9e52875d3a8a216248d70b"}]},"branch":"refs/heads/stable/zed"},"0f2b884e3d430f14ef8ef1a50ca2c6b52cd2528e":{"kind":"REWORK","_number":2,"created":"2023-03-15 14:21:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/877409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/09/877409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/09/877409/2"}}},"commit":{"parents":[{"commit":"9802342616d3954832f14780ec2b74e0e59de932","subject":"Merge \"Wipe Agent Token when cleaning timeout occcurs\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9802342616d3954832f14780ec2b74e0e59de932"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-03-06 18:48:07.000000000","tz":-480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-15 14:21:18.000000000","tz":0},"subject":"Fix online upgrades for Bios/Traits","message":"Fix online upgrades for Bios/Traits\n\n... And tags, but nobody uses tags since it is not available\nvia the API.\n\nAnyhow, the online upgrade code was written under the assumption\nthat *all* tables had an \"id\" column. This is not always true\nin the ironic data model for tables which started as pure extensions\nof the Nodes table, and fails in particular when:\n\n1) A database row has data stored in an ealier version of the object\n2) That same object gets a version upgrade.\n\nIn the case which discovered this, BIOSSetting was added at version\n1.0, and later updated to include additional fields which incremented\nthe version to 1.1. When the upgrade went to evaluate and iterate\nthrough the fields, the command failed because the table was designed\naround \"node_id\" instead of \"id\".\n\nStory: 2010632\nTask: 47590\n\nChange-Id: I7bec6cfacb9d1558bc514c07386583436759f4df\n(cherry picked from commit cbe5f86ce7c100f834701cf43f7b9b1bfdfb9ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0f2b884e3d430f14ef8ef1a50ca2c6b52cd2528e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0f2b884e3d430f14ef8ef1a50ca2c6b52cd2528e"}]},"branch":"refs/heads/stable/zed"},"2b0a5575ce30e373f709388c2697fd9775b2850b":{"kind":"REWORK","_number":3,"created":"2023-03-15 15:41:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/877409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/09/877409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/09/877409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/09/877409/3"}}},"commit":{"parents":[{"commit":"9802342616d3954832f14780ec2b74e0e59de932","subject":"Merge \"Wipe Agent Token when cleaning timeout occcurs\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9802342616d3954832f14780ec2b74e0e59de932"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-03-06 18:48:07.000000000","tz":-480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-15 15:41:47.000000000","tz":0},"subject":"Fix online upgrades for Bios/Traits","message":"Fix online upgrades for Bios/Traits\n\n... And tags, but nobody uses tags since it is not available\nvia the API.\n\nAnyhow, the online upgrade code was written under the assumption\nthat *all* tables had an \"id\" column. This is not always true\nin the ironic data model for tables which started as pure extensions\nof the Nodes table, and fails in particular when:\n\n1) A database row has data stored in an ealier version of the object\n2) That same object gets a version upgrade.\n\nIn the case which discovered this, BIOSSetting was added at version\n1.0, and later updated to include additional fields which incremented\nthe version to 1.1. When the upgrade went to evaluate and iterate\nthrough the fields, the command failed because the table was designed\naround \"node_id\" instead of \"id\".\n\nStory: 2010632\nTask: 47590\n\nChange-Id: I7bec6cfacb9d1558bc514c07386583436759f4df\n(cherry picked from commit cbe5f86ce7c100f834701cf43f7b9b1bfdfb9ea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2b0a5575ce30e373f709388c2697fd9775b2850b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2b0a5575ce30e373f709388c2697fd9775b2850b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
