)]}'
{"id":"openstack%2Fironic~884413","triplet_id":"openstack%2Fironic~master~I416bba720af542abd9549ebc853d896a125c5890","project":"openstack/ironic","branch":"master","topic":"test-metal3-integration-job","hashtags":[],"change_id":"I416bba720af542abd9549ebc853d896a125c5890","subject":"[DNM] troubleshoot metal3-integration job","status":"ABANDONED","created":"2023-05-25 15:32:08.000000000","updated":"2023-08-29 08:41:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4ed7eb92acf40adec347ce3da8dd61895f52b0d","_number":884413,"virtual_id_number":884413,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-25 17:00:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 15:47:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 17:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef0c6c93bb888dc5ce431f0f9f1cf614927027e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-05-25 15:32:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc18f56e2ffbc044448f3146f6bbda7d3954dbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca9d54e8c2e447bdaf1b5252322390f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9afe515a528149c3bd969f24b8b2a17a : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a1670cc57474cadb848bd906f614e2a : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edaeeb291ee5470fbb21af2dd28cc86f : SUCCESS in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7e42f82b8c45ef274b4933e0bd953441f9aa27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:00:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da585b96fb484a45a6aa3e207a0ab023\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f4252f7980a4ec98c49e26bb6e49ced : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963aecb86f1042168d96f1b75f63a160 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0055cfc64b89425999dc34fcca368943 : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb6497555fee4ae49e98af3cdd497a8d : SUCCESS in 9m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fa35c031ff14646aaeba2b5e3beeef6 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955979665bea4f308719e78779f4b8ab : SUCCESS in 14m 26s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/fb3c38b6ace34c21b3029885a4da0104 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a4ed7eb92acf40adec347ce3da8dd61895f52b0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-08-29 08:41:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37347ae041299714c4cc0a65f40478fa3ee5a58e","revisions":{"37347ae041299714c4cc0a65f40478fa3ee5a58e":{"kind":"REWORK","_number":1,"created":"2023-05-25 15:32:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/13/884413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/884413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/884413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/884413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/884413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/884413/1"}}},"commit":{"parents":[{"commit":"443cbdebce18fefa53271b2fdf522ac86e03e6a6","subject":"Merge \" Add DB model for Firmware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/443cbdebce18fefa53271b2fdf522ac86e03e6a6"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-05-25 15:31:48.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-05-25 15:31:48.000000000","tz":120},"subject":"[DNM] troubleshoot metal3-integration job","message":"[DNM] troubleshoot metal3-integration job\n\nChange-Id: I416bba720af542abd9549ebc853d896a125c5890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/37347ae041299714c4cc0a65f40478fa3ee5a58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/37347ae041299714c4cc0a65f40478fa3ee5a58e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
