)]}'
{"id":"openstack%2Fironic~884608","triplet_id":"openstack%2Fironic~master~Ic3df86f395601742d2fea2bcde62f7547067d8e4","project":"openstack/ironic","branch":"master","topic":"handle-duplicate-inventories-db","attention_set":{},"removed_from_attention_set":{"35929":{"account":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"last_update":"2023-06-14 14:08:04.000000000","reason":"\u003cGERRIT_ACCOUNT_35929\u003e replied on the change","reason_account":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-06-14 11:46:20.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":[],"change_id":"Ic3df86f395601742d2fea2bcde62f7547067d8e4","subject":"Handle duplicate node inventory entries per node","status":"MERGED","created":"2023-06-05 05:38:41.000000000","updated":"2023-06-14 16:35:32.000000000","submitted":"2023-06-14 16:33:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884608-handle-duplicate-inventories-db","meta_rev_id":"c6994ee89887ce7fce770aca8c6cab60a732fd79","_number":884608,"virtual_id_number":884608,"owner":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-14 16:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-06-14 12:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-14 11:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-06-14 12:35:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 05:40:57.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-06-05 06:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-05 08:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-07 08:12:06.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2023-06-14 12:35:19.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"f07fff2269742c0cd9ae89d59c5a5ba722e0d74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-05 05:38:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec773951675c8d303ade6621e624ded0e22ee22","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-06-05 05:40:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct : NODE_FAILURE in 0s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"224684a5caf23fd86a13ce66c4d7c253e70733ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-05 05:45:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d89ac5e9a0fee30a7bd00092c48dfcf8435794c3","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-06-05 05:50:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct : NODE_FAILURE in 0s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"93864855902eac8b5f4657e3f3a77f70021583ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-05 06:18:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a1cec7b3abc13e707d1d61b7683baca0e6a91ed","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-06-05 06:21:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct : NODE_FAILURE in 0s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"03371631925690b79e7737d4664f395019722258","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 06:27:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070ffa7568844af9a21b536fd61a21bd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4eb18ee0fdc94b8ca34acd683b76740d : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b74e1500f09048a9958a2e97dfb5de45 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0483798a72e2498c8e0f4d96bc0708ee : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7cbaf494e2e68c15d7442963ff549bf5b084cf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 08:28:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd107e7ed337477b96affd3a30938475\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03d357c4ca0c4c1ab7b790f42ebc8848 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35700e38fe2c41ca8ca0c4ec95187f7a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28fe1ac4ac1f4f58a76a31a0c3cf8603 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15317c5f30c247b2ba108f8803eb3d3a : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae21525d30dc41fb9393ba943730311c : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54be46a20e114d1ab4cf07b8cf5ece03 : SUCCESS in 14m 10s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/694935dfd16b4ce4bb114ed20ddf4acb : SUCCESS in 10m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9dfd5953b4c446aabe6a1072b1abaaf9 : SUCCESS in 9m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b65aa01f46a7418fa21b81707c9ccee9 : SUCCESS in 31m 49s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/81817c1ec03e418c91da4ba41f6bca45 : SUCCESS in 24m 02s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/223e65e2257f4ad381fd460fdf1923b0 : SUCCESS in 56m 05s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/638f2d9bd151474690c4041a33c33dd4 : SUCCESS in 1h 16m 32s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/398d947f4b6346a78dc56d30b2d371aa : SUCCESS in 1h 13m 32s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/9d48eb1e8ad5475d8c596db980adfe6b : SUCCESS in 42m 46s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cca6af7b16dc402aa5695232e5c9ffa6 : SUCCESS in 38m 29s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/409d177d3cd34e7084837d45ae462b78 : SUCCESS in 56m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/33dc4cb3ae9240e98776889b0d469713 : SUCCESS in 48m 03s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/4de07e09c20841749f7c6d91a4e712f0 : SUCCESS in 43m 56s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d79b157291bc4cd9adf825544ce90bde : SUCCESS in 2h 01m 59s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/93f07dcb848c447f9129de8741843272 : SUCCESS in 1h 09m 59s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4378d4fb7c0443d5a3c33a52344c147c : SUCCESS in 48m 18s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7fa95976d0684d66b2fee83c7197742a : SUCCESS in 56m 35s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/b381759b03c1435f96cab46450a902b9 : SUCCESS in 1h 22m 04s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/4f0ec4a3735f4119bab6a48409b72a18 : SUCCESS in 40m 21s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/3943a397205d4e08bf831af6a4b0a7cb : SUCCESS in 5m 21s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/77a15ec74cff468b8a2d9976c9a974b0 : SUCCESS in 56m 18s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0ab01eb34cb04235a209669717b19e22 : FAILURE in 56m 37s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/939c787d2a7b4768acbbfbfb2888db36 : SUCCESS in 36m 59s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/738a732a2397495594d8b399d8e4f329 : SUCCESS in 57m 32s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/accb0d4b3c96434c9b119213e00fe1da : SUCCESS in 34m 58s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/680824328b0f4395a1b1ba590f3db529 : SUCCESS in 43m 47s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1734fdd4d70485781727086a0747b8e : FAILURE in 8m 14s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/4da464b733a747b194bd16613b914ba4 : FAILURE in 8m 38s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/1f30e85a91ef4ff19c58a7f158777054 : SUCCESS in 42m 30s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/c43af2e466b142d4a9103f866c3a0b36 : FAILURE in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"95c77ddffb28c3aeacdbc6e5ccb5915d2a0b4b68","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-06-07 08:12:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"002de574d888203a9baa34d60d22256350500fb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-07 10:28:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c01910d84ced68fd1efab568ce35b22ad0b2ea5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:37:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe1a22862624bbcb497a657816a41ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ffeea255c7e461a8ab7b59dd7a9d092 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8137a64feeb34551ba8a944c6ed844d1 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7025bd4023d437cba3ea5ee1ec86512 : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3160120167f321336faf8d9367b295d5e6050d92","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-06-07 11:15:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/4/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/01410d1/01410d1.html : SUCCESS in 45m 23s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/4/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/0104c8e/0104c8e.html : SUCCESS in 16m 41s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/4/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/fb9a726/fb9a726.html : SUCCESS in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"64c5213c521af7aa7f1ae33a141b0c32d81242bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:51:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a40ae76830e04aed88270dcf3c355b6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/039c4c6c12cf4bd7a4eb9d6e2353fe36 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b22a2175c1746a698f0e1894ee207b8 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29bbf34775554a32b517b27de4e9e799 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0df3cc6ec51c4e43b99b6284436f0a3d : SUCCESS in 9m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ac066474b77492cbe7e18614b8847f3 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60c6304855745438e21f699a691cada : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f94d451946f04ad4b58ed1210cd6c8eb : FAILURE in 7m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f6cf9bdd34ff41d6b1ab57930e018603 : SUCCESS in 10m 11s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3e0288f82fdb4b52a9a6c180b20ead12 : SUCCESS in 8m 01s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/cb67d629631149d0ae86dd25db9f9277 : SUCCESS in 32m 05s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/8c389db063974f98b5195ef84a3332e1 : SUCCESS in 25m 52s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/9ab222f8d121449cbe2a860936582c68 : SUCCESS in 52m 01s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c406b4e77a0642a2aa19d92bc120a442 : SUCCESS in 53m 43s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a22d19f6fc474fe7824a3af444492758 : SUCCESS in 43m 17s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/61c5c0acc79447f08330e45b8c634b24 : SUCCESS in 41m 11s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/92daa8a0cf7245938f329b4d9d7ecad1 : SUCCESS in 38m 22s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0081157ebf9149998d05cc4e1f68afbc : SUCCESS in 48m 18s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3f2e0638f69f40f08108ca62bac18f2d : SUCCESS in 40m 09s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/505613a3b86f44fdb6155fc40ee8372a : SUCCESS in 49m 40s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/34c7d5426467455fb3b39a173c78bd4c : SUCCESS in 1h 21m 32s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b07c2fd41bca49dbab7137f6e58e6beb : SUCCESS in 51m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0054ad97fe1a473e971e99918a91f545 : SUCCESS in 33m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/93cde2d5d9dd4458b1f1c59668e2be2f : SUCCESS in 46m 11s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/547597a80fe04daf8c50eb36386e7978 : SUCCESS in 1h 14m 37s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/57a5a19c43174a4d9191650d9e5b4920 : SUCCESS in 38m 11s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/fc9ca2a569724be0908844d110e0f94a : SUCCESS in 4m 38s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/622da55a1f6d41f393066ab3c36cbde4 : FAILURE in 1h 12m 59s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/d463979fe9c24b5f8929914769bd2b44 : FAILURE in 46m 47s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/92d7885d69fb4114814e4683b33a9148 : SUCCESS in 36m 41s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/20aec0968237402f85eacfbc98bd2b82 : SUCCESS in 1h 02m 22s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4efcd13f2a7b43df9d5b9b01d8b194bb : SUCCESS in 50m 37s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5b3f379d716348f285d6b39bfc7f08c3 : SUCCESS in 1h 01m 57s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34f7867a03714b61b8b1cbad103b3f4f : FAILURE in 7m 38s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/523ec4928e7849168c5bb4e8bf5ac403 : FAILURE in 12m 05s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/1da803b6355a44a6a11e4e4035dbc551 : SUCCESS in 46m 58s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/b87266c74ca046f6aa7fee8d54381672 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28745504697f4c9da04eb39b92d3514717fd4c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-07 12:09:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5d2edbd57a2df156f3a05b07f79365fa183eddd6","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-06-07 12:54:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/5/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/b35875b/b35875b.html : SUCCESS in 40m 54s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/5/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/419e5d6/419e5d6.html : SUCCESS in 17m 00s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/08/884608/5/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/fb55850/fb55850.html : SUCCESS in 40m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2d9bfb99673bce82862e85a0a49d5fa47ab785bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 13:12:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443b291f08f046c5bc98e2c6cac1218f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a6872cd1c914f2fb06a4eaa07f4d9f8 : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bb3abbe5d874ca2babdcd91c80873a2 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96b19427c32b484c827427fa048d5aff : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"061006bec24495b16509fe58df57b518c093ca6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 13:49:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7a848d676ac41fb8d47bd307ae3cbef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e48cfafbee04239998953259c992ba1 : TIMED_OUT in 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d27f8f74d94a86b758665653012ada : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/850fcd3e00334a1f94f7dbe0dcb04d9b : TIMED_OUT in 40m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbe1973d2e5043ed8bfdf022481b9192 : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0f32b89d7044c84b8ede863af72db69 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c441d7c3c42a494e86abe6cf1d2eb766 : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5e20a311dd46ccb6bf63676235543f : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/34667b31d5a448dba94bf4720989de08 : SUCCESS in 9m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1574b4f977584e3ca7cb05e21961347e : SUCCESS in 7m 17s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/a585341209a44439ad2bf8bfa9c5db22 : SUCCESS in 22m 58s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/f92f9b4420524df4b69d835c167f6022 : SUCCESS in 20m 22s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ecb187b2ad89486d9c4e8be81d9a9ef6 : SUCCESS in 50m 23s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe2c8424a53240b6abd62a2c247f2883 : SUCCESS in 1h 08m 45s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/645f2ced94ed46eabe580656488d6e1f : SUCCESS in 1h 00m 40s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/4f8064ac4ebe4b7fbfb595d57bb26a70 : SUCCESS in 51m 44s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3756c084224248f59da3c5620ded03ac : SUCCESS in 37m 23s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/595ea3e043244d1c810375f8599a9396 : SUCCESS in 50m 28s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8a7eac53faac4c60aa946cc139ea003b : SUCCESS in 40m 35s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/02f233e6237f447b8efea2dc45cfeef2 : SUCCESS in 48m 49s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/800566b5523a4acaa118c9e7d25bdfcd : SUCCESS in 1h 38m 02s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/47eb503bac3c42bcb87dfc10be302d90 : SUCCESS in 50m 13s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/b9a5c6dc57d14419aca4e911f5517ae7 : SUCCESS in 34m 22s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7f1623c353ca4756bc860faa24a9d613 : SUCCESS in 46m 53s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/cd825b66a600470c8223b75ad2aac041 : SUCCESS in 1h 20m 00s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/1077d9d04b1e49e8bd9a9055c2540ff9 : SUCCESS in 38m 19s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/788d50f756eb405db02a2e110e8a1b3c : SUCCESS in 5m 50s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a38aea0f43a84c13a0a9fe747272ab99 : SUCCESS in 1h 01m 15s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/1b2cd476d54741978ad8ce3788c01cdd : FAILURE in 54m 55s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/2f375d461f7f48758158aad4b1687ade : SUCCESS in 44m 40s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/751a34660b1d45acae067a4daaa748a1 : SUCCESS in 53m 49s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/b2edd378e05540a3b686474659fdc2f0 : SUCCESS in 39m 39s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/d7083d4ad37b4fe8821c9fcabff7b42e : SUCCESS in 50m 34s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2eb3f057f3e4ae888460f181daaba97 : FAILURE in 9m 07s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/600f376c68e748508763e4489728b642 : FAILURE in 10m 06s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/1a7dcabeb60a4ad398ef40cedf4b79b9 : SUCCESS in 50m 03s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/db48845ff8cd4f269e896132c59e6204 : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab24bed1e38977ec8470007e5cda6767eebd2e71","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-07 15:45:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e6b54f5ecee15c52d392428629d4e855460c3ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 16:48:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9089efc119b44f582bc920680d85e9e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/064d6de0726146849e0bf666435bbfe4 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5243897255cd48e6be92d4a630f46577 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93b6b5b9eb53494c96afd6faab2fd69e : TIMED_OUT in 1h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"79ab49e456a156c316d646fa2c8fa933dc7c0937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 17:05:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8f53c601e14b9d9e6b5ca475fe9f1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a509e996ed8c48d8be20cc92fd2282b2 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210441c7b8be49bf95ef9706f642df44 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a6c3096d7ab47318c0bfd8aab74e3e5 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8dfa8ae6bf4dd180dff6cc2adf4d39 : SUCCESS in 8m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2099e68095b4c9087fe51cfcd39c581 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46126a5c003e4741bd3aca8c78d9bf92 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c440350e3c141eaad0bb0ee7a9dc48b : SUCCESS in 7m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ab2a0d07b28c4c8981d75abf9cbeb392 : SUCCESS in 10m 56s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1df71f4dbf084af3840732ca9b0337e2 : SUCCESS in 10m 52s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0b09b288ac3747fa944d965e1dc37ed8 : SUCCESS in 30m 37s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a9dc5027de0b4a879743090feda40332 : SUCCESS in 36m 44s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/3aa2f81b3a4b4604ab8a45f3b2994d82 : SUCCESS in 1h 04m 06s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f48190eb6584b4aad09ac2151ba74b9 : SUCCESS in 1h 04m 26s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d725a2235abf423484aacea2ec51f791 : SUCCESS in 40m 42s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/82b7098f8cb149b9850b01bd8afccef0 : SUCCESS in 46m 05s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2213417fd96843a795669ce6960f41b7 : SUCCESS in 43m 12s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/78d1065098cd4e9f9e3fd36324e2de0c : SUCCESS in 45m 36s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/8a5642ae95c744288d4ec68c13d72da0 : SUCCESS in 32m 27s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/e3a79690e2d747139ac6d15b997e162e : SUCCESS in 45m 18s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/db29d187368543ec98aed29db7af0853 : SUCCESS in 1h 16m 07s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/82cee461a82f4155bc096d2e7f774a47 : SUCCESS in 52m 01s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7d8f88e9a5c04a76ad2372293efa9219 : SUCCESS in 36m 29s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/5d2333097b1342ddb4e6c5ea304ae641 : SUCCESS in 38m 43s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/3ef7b4f39b67491088951f2d2c58c7de : SUCCESS in 1h 06m 51s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/7a5154b48469417da6eb84308aaa9c9e : SUCCESS in 37m 20s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/dc61ac70636c4269b0ae280c6e9ca17f : SUCCESS in 6m 07s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7965a7c0fe9049bbb4d25b76f76467df : SUCCESS in 45m 52s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/73df4b216de84b03a0d65784a5a3b796 : FAILURE in 47m 44s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4fa7af3525df480ca825fe0e2dd324d8 : SUCCESS in 33m 52s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e2cba16373f14014865f7865a7b4b5d2 : SUCCESS in 56m 24s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/c67c987c77c345ecba3d2fdfacc1e08d : SUCCESS in 54m 04s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/f3cc8f1b2201407fb294646879ed9e0b : SUCCESS in 42m 11s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1aa3a2685f74290a44576154dbb7a03 : FAILURE in 8m 22s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/923c3e5ec93847c380ef97656d4985f0 : FAILURE in 11m 57s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/eb8bb88fc77a444ba86b7b312b8b7fe4 : SUCCESS in 56m 00s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/2d131434202d4b2fa49e0f2c55e46737 : FAILURE in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e687db12d7128c2b6efbf59768d0891933595eff","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-06-14 11:46:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37edb9b632515a7f11d5c61837651c5be93385ec","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-06-14 12:35:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"48e2581aabb6cb2ececb81707b5c2d1532b6e504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:35:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2315600a248f864ace51b47200c7fab092061e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 13:44:17.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f34ed5f642d4b32961b1ec42f25128e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888307b91a16491781b8bfd8a4ab1805 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c35d6f84cb9044f7896b623fb73bf707 : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fd0466374e243b1bb5f45b0d6806aed : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c77ad51621c04032930d6999993bbec7 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164700de0cce4d6fb45b584c7e07048b : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e8dded20f94a9b82416c91c299ad8c : SUCCESS in 5m 19s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d217f7b02eb54e9bb4f6b83bf6aaeace : SUCCESS in 10m 33s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/acaa5c94c1fa4430bab755a22e966c8c : SUCCESS in 19m 35s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/15fbf24e4fcb443e827bf99f29d4e481 : SUCCESS in 55m 43s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/53b8263aa4be4489874a520419b9d09f : FAILURE in 49m 25s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/c632cc07f6cb48ffbf0532c5354597ba : SUCCESS in 48m 11s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/34341055cb7348c2b72e3074690d7143 : SUCCESS in 44m 11s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/56f08cadb9304b6d96d3bc0c166dc86d : SUCCESS in 54m 03s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/3d8d0c1f66f5430e84a580bfd747c02b : SUCCESS in 28m 57s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/934ad5627f1c407cac20a1cd33de113b : SUCCESS in 57m 29s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/159c95d09bb54a7b98e546b4a0d46ab5 : SUCCESS in 1h 03m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/77b74e3d286b453e8d65cf910ce7c2c9 : SUCCESS in 46m 52s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/79561a029db44a40a6700568c29006e3 : SUCCESS in 37m 00s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/4208c0d39822419fb4cc9aa233142694 : SUCCESS in 1h 07m 47s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/08b2bf1e34eb4cda838a46b6a0d180aa : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":5},{"id":"63421b856b128662527d89986457f6d4ee62801e","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2023-06-14 14:08:04.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9d2c124dcebbccd497cceba04e094912b15fc234","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 14:23:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cec07716354f4b39a08965fa92646a2a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/501c3474df0b480bb466b7a69922ed52 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f5c4419072f40abb3cd46c08ab8e23d : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2b5b5cb350c473e8473e4aea84d6ae8 : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4d2ec2fc88046fffe5b3d1edd1341da0e7caa8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:31:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64168dd3f35c455b8292002766cda80e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0dcba5263548e3a088959cde5de8b8 : SUCCESS in 48m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e1fdf4f5e7416c964768a19ac8a946 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf6bb1864f14f7490bb390f4a25c588 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b732b3c6ef3a4606b919612075927d4e : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e4be3f1f56a40a8b34f2f6a6a7e5265 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e775d67cce8d47b4a57065025aee187f : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a340fdc4509429db9321bb76ecced5b : SUCCESS in 6m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/4db80709dbf141e883ea0190e839ec0c : SUCCESS in 9m 26s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/403cf4df58804574a5f28cb417005b27 : SUCCESS in 8m 52s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/0de4462e39dd478c96c03b90b42089ce : SUCCESS in 17m 58s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/8a3c3475a42c400d83ce59fc8ce23bba : SUCCESS in 25m 08s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/addbb2e3d3154d48b5fd4c2dfc0d4b5f : SUCCESS in 1h 17m 51s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa6ef1f4bafd4c90b7c8e0902ee067de : SUCCESS in 51m 29s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/1e59debab61449b581a8e5c13410f09d : SUCCESS in 45m 01s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/a6255825e8c14413bb0311f2cac82de5 : SUCCESS in 54m 12s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/df03f6f550574b6fa13ffccb48f06bbd : SUCCESS in 46m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/99257e3b69bc4e2a8dacc278b4fc5147 : SUCCESS in 55m 38s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/bd70efffef2b413d8f6b2e5b0ffc6b37 : SUCCESS in 40m 56s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a7c1f8a670e04aab89df29f6db59fe1c : SUCCESS in 48m 51s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/aef5b5ac8ac2431eb0b5a7d5371848fd : FAILURE in 58m 05s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/d542af54dc394530b9891457a52082d7 : SUCCESS in 1h 06m 17s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/4dfc120562a24b2faeffa66589ce7f65 : SUCCESS in 38m 53s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/abbe9d52d4b447bb8474107bcafff99a : SUCCESS in 44m 00s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/b36ca42591af4d27a6bb90fd6477b1c0 : SUCCESS in 1h 11m 15s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/07c6896c67c5429f9d9af6b82cd454d1 : SUCCESS in 31m 16s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/2a400e9f05b945e5beef7a26597e8bff : SUCCESS in 5m 22s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/96f2c652777d47379fb7d8de8348c663 : SUCCESS in 45m 11s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/aa8fc63f10004312a7223ac5f777c595 : FAILURE in 45m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/79d5f0a915a44bb79447e8788e66eabe : SUCCESS in 35m 04s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/e221d71fdd5142a48fc886ceecfdc37c : FAILURE in 1h 07m 26s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/6381593f7f6547c7a41502d8d7767e06 : SUCCESS in 35m 24s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e6e0eaf721db4e108415fc2c59ca91b6 : SUCCESS in 1h 07m 20s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47bab35a88ab48279ecab936bde1fe52 : FAILURE in 10m 15s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/68bad2052b014f55bccaa2ff89e7781d : FAILURE in 10m 08s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/2d2452762645421fac391c10d0d384df : SUCCESS in 49m 43s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/6c16dd605dfe40a7a421ef8bc4e0d4cf : FAILURE in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c3bff9ae2830d0f896cb903a45fcdc09302ca178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:32:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"96cf3e025bbfa8a0b06d575d556c4461c6f70ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 16:33:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c8296749a9415787eb51d189117046\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2ed440cba44a76942f05a7cfe20243 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a4c7e71310844eb8c9ba8862bfe1122 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ded9006131a472c9d81e3bcdfa6f3bf : SUCCESS in 8m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17538fab75f847d4bdece030d97450a0 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8949bbb4884487a42f29df0bcbb804 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cbbff301c9242eb8d2c62c75ea572c7 : SUCCESS in 6m 49s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d38e2e28d03646938e371237ed3a56ca : SUCCESS in 9m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c018e22ee25f4d66b56e8b0e45380cf3 : SUCCESS in 20m 50s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0fc3cdb58b3141d2a88724e7be802e84 : SUCCESS in 59m 05s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/91fefad1e90048f09a99999c03172b08 : SUCCESS in 55m 48s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/d6610a49aedc43f78806fe81593cb176 : SUCCESS in 38m 15s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/411f1189ea21458795072e383d4d7a63 : SUCCESS in 41m 41s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5051eacb653f4c74b6bd1c75cc02f77d : SUCCESS in 56m 22s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/41f71e2115b74550b218067f78e3aa06 : SUCCESS in 41m 31s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/06ae94f4854f4271ac64bb63c8dbbbe0 : SUCCESS in 54m 51s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9fb30a13c9924241b9b40b31b9c460de : SUCCESS in 57m 19s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/7f8bcc45ad05442ea0bc925de5440700 : SUCCESS in 34m 01s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/3be0bfc11993467689b4a3642346a305 : SUCCESS in 44m 28s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/9773c5d9614641119c28a5bebe9efac1 : SUCCESS in 59m 24s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/1ba1e06ff9f8476babb985da6afb49c8 : SUCCESS in 47m 27s","accounts_in_message":[],"_revision_number":5},{"id":"acb960d8f60356827d923aee72eba7d3aef655b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 16:33:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c6994ee89887ce7fce770aca8c6cab60a732fd79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 16:35:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6caac753b6d4801b58c4a98eae22d2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae617c48faa47f0bb6c6c08ee96bfc2 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7dd005c6061495d8d65ec9d1b67856f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fa2d6685f373168a9dfe4cd1f7d83af9d87a025c","revisions":{"bfe0e8e2fa74c5c74edfbb25352ea23e16f73418":{"kind":"REWORK","_number":1,"created":"2023-06-05 05:38:41.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/08/884608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/08/884608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/08/884608/1"}}},"commit":{"parents":[{"commit":"964a82db18dc322503b49274688ee5ebc00dfb87","subject":"Merge \"Add to Redfish hardware inventory collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/964a82db18dc322503b49274688ee5ebc00dfb87"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"subject":"Handle duplicate node inventory entries per node","message":"Handle duplicate node inventory entries per node\n\nWhen a node is inspected more than one time and the database is\nconfigured as a storage backend, a new entry is made in the database\nfor each inspection result (node inventory). This patch handles this\nbehaviour as follows:\nBy deleting previous inventory entries for the same node before adding\n a new entry in the database.\nBy retrieving the most recent node inventory from the database when the\ndatabase is queried.\n\nChange-Id: Ic3df86f395601742d2fea2bcde62f7547067d8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bfe0e8e2fa74c5c74edfbb25352ea23e16f73418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bfe0e8e2fa74c5c74edfbb25352ea23e16f73418"}]},"branch":"refs/heads/master"},"026ac092875d4cc81fbf21943876b497b1d6efba":{"kind":"REWORK","_number":2,"created":"2023-06-05 05:45:44.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/08/884608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/08/884608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/08/884608/2"}}},"commit":{"parents":[{"commit":"964a82db18dc322503b49274688ee5ebc00dfb87","subject":"Merge \"Add to Redfish hardware inventory collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/964a82db18dc322503b49274688ee5ebc00dfb87"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:45:34.000000000","tz":-240},"subject":"Handle duplicate node inventory entries per node","message":"Handle duplicate node inventory entries per node\n\nWhen a node is inspected more than one time and the database is\nconfigured as a storage backend, a new entry is made in the database\nfor each inspection result (node inventory). This patch handles this\nbehaviour as follows:\nBy deleting previous inventory entries for the same node before adding\n a new entry in the database.\nBy retrieving the most recent node inventory from the database when the\ndatabase is queried.\n\nChange-Id: Ic3df86f395601742d2fea2bcde62f7547067d8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/026ac092875d4cc81fbf21943876b497b1d6efba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/026ac092875d4cc81fbf21943876b497b1d6efba"}]},"branch":"refs/heads/master"},"4808fae1acb8d578f1a52ac542e8dbbcd42f39c1":{"kind":"REWORK","_number":3,"created":"2023-06-05 06:18:20.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/08/884608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/08/884608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/08/884608/3"}}},"commit":{"parents":[{"commit":"964a82db18dc322503b49274688ee5ebc00dfb87","subject":"Merge \"Add to Redfish hardware inventory collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/964a82db18dc322503b49274688ee5ebc00dfb87"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 06:17:51.000000000","tz":-240},"subject":"Handle duplicate node inventory entries per node","message":"Handle duplicate node inventory entries per node\n\nWhen a node is inspected more than one time and the database is\nconfigured as a storage backend, a new entry is made in the database\nfor each inspection result (node inventory). This patch handles this\nbehaviour as follows:\nBy deleting previous inventory entries for the same node before adding\n a new entry in the database.\nBy retrieving the most recent node inventory from the database when the\ndatabase is queried.\n\nChange-Id: Ic3df86f395601742d2fea2bcde62f7547067d8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4808fae1acb8d578f1a52ac542e8dbbcd42f39c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4808fae1acb8d578f1a52ac542e8dbbcd42f39c1"}]},"branch":"refs/heads/master"},"dbfd2388080a741d196aeec81d2b655ea17c6727":{"kind":"REWORK","_number":4,"created":"2023-06-07 10:28:47.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/08/884608/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/08/884608/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/08/884608/4"}}},"commit":{"parents":[{"commit":"964a82db18dc322503b49274688ee5ebc00dfb87","subject":"Merge \"Add to Redfish hardware inventory collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/964a82db18dc322503b49274688ee5ebc00dfb87"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-07 10:28:15.000000000","tz":-240},"subject":"Handle duplicate node inventory entries per node","message":"Handle duplicate node inventory entries per node\n\nWhen a node is inspected more than one time and the database is\nconfigured as a storage backend, a new entry is made in the database\nfor each inspection result (node inventory). This patch handles this\nbehaviour as follows:\nBy deleting previous inventory entries for the same node before adding\n a new entry in the database.\nBy retrieving the most recent node inventory from the database when the\ndatabase is queried.\n\nChange-Id: Ic3df86f395601742d2fea2bcde62f7547067d8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/dbfd2388080a741d196aeec81d2b655ea17c6727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/dbfd2388080a741d196aeec81d2b655ea17c6727"}]},"branch":"refs/heads/master"},"fa2d6685f373168a9dfe4cd1f7d83af9d87a025c":{"kind":"REWORK","_number":5,"created":"2023-06-07 12:09:47.000000000","uploader":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"ref":"refs/changes/08/884608/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/08/884608/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/08/884608/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/08/884608/5"}}},"commit":{"parents":[{"commit":"964a82db18dc322503b49274688ee5ebc00dfb87","subject":"Merge \"Add to Redfish hardware inventory collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/964a82db18dc322503b49274688ee5ebc00dfb87"}]}],"author":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-05 05:33:32.000000000","tz":-240},"committer":{"name":"Mahnoor Asghar","email":"masghar@redhat.com","date":"2023-06-07 12:08:37.000000000","tz":-240},"subject":"Handle duplicate node inventory entries per node","message":"Handle duplicate node inventory entries per node\n\nWhen a node is inspected more than one time and the database is\nconfigured as a storage backend, a new entry is made in the database\nfor each inspection result (node inventory). This patch handles this\nbehaviour as follows:\nBy deleting previous inventory entries for the same node before adding\n a new entry in the database.\nBy retrieving the most recent node inventory from the database when the\ndatabase is queried.\n\nChange-Id: Ic3df86f395601742d2fea2bcde62f7547067d8e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fa2d6685f373168a9dfe4cd1f7d83af9d87a025c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/fa2d6685f373168a9dfe4cd1f7d83af9d87a025c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
