)]}'
{"id":"openstack%2Fironic~887166","triplet_id":"openstack%2Fironic~master~I5a65accfb7e2033690b2934d874141db7f4bf383","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-06-29 16:12:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a65accfb7e2033690b2934d874141db7f4bf383","subject":"Execute tests by class, not randomly","status":"MERGED","created":"2023-06-28 13:30:19.000000000","updated":"2023-06-29 16:14:04.000000000","submitted":"2023-06-29 16:12:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887166","meta_rev_id":"95d9bf82b390359b111aece97ab7b22f380ab0c1","_number":887166,"virtual_id_number":887166,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 16:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":2,"date":"2023-06-29 02:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2023-06-29 15:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":1,"date":"2023-06-29 15:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 13:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-28 13:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 02:16:34.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2023-06-29 07:23:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2023-06-29 15:52:51.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"1b6dded017e445abcabb1013b894aeb1c6693864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-06-28 13:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efcad421e49763a247b31788865edf07b181cbda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:43:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25393ffdf5d743fb98e8c46abd8363be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf5f6b66fb3940aea76a8cae0db76774 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32ed814f73b54428842ad9dee1cf1376 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a840ff053854b7c8ec4ce78a3442932 : NODE_FAILURE Node request 300-0021565334 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bbf0fc5204291ac48ca152e7ddd22bf1ca82fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:54:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9ed4ad1e65349b6986752fbe57e139b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/76f9a622a7d649fdb506155d2a911a27 : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a39f72a877de4b0aab7e6c40941e48f1 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4f6baf4ecd477d938fab5d6a44b4ea : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f648a8c2fe40d19ec434d213c267c1 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cacc7c0dd444a89a2a4d976837712a7 : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4087e979a0a34eec8e3a5d2c0ac14d66 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86b655ed14e40bc9fda5cb106f6aa5f : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c27cc8fda6451aacf9ae8dbb207e1e : SUCCESS in 7m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ecf94286b00646d1a44720ee73791603 : FAILURE in 17m 33s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b35412f161884000b21a471ab63d8c75 : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6db97a92bf1d4840669746dfc0300edc0065478e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-06-28 16:40:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d969e32af1bb4784d993aff0b6ea2f775e0f15eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 16:45:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7314a22027cd41529761bea7b63f9e6e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e2ed157b3694dde87bada3e82550d67 : NODE_FAILURE Node request 300-0021566854 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7217dd5b7a64c3a8bd32929236994ae : NODE_FAILURE Node request 300-0021566855 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b252ae74b1af409896ee7a9c0f82e943 : NODE_FAILURE Node request 300-0021566856 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"226959dd2981ca4ff750211fda7d591bc06115a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:37:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0649e483d26e42faa17edc55f218f6e6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/321e29afa26547bebe8969260ac8da94 : SUCCESS in 8m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5abfe428fc84ce29023a3223aa3786e : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330018e933cb47e7a4a05fd1b45a00db : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d028e4d7bb4af294a052ede17fd888 : FAILURE in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98cffe12fca64ac4bdf3feef69034187 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b458e5c628854012a3130ac9e29bbd10 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15e36daf6c94450ad5a40bc9b3edde3 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac618ae9d8524a198605c6abe33d7747 : SUCCESS in 6m 56s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/6cab9bb5c7084289adbf2a593c258c90 : SUCCESS in 8m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2a679f55b47645dda81a2fa899babf5a : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26365dc4aba709bc024a0461e2414eb3a025156d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-06-28 22:45:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e6f2c1eb788c8468e805b611c5931b587f61d560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 23:02:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7519f8654d8e497d8bd37bebce1592fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/09db0c220d8043b181831b3a6ab9e662 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5deae18ee0a1466cbee8033364b34f05 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5fe643198f4d098544995311905d79 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d02c8b1926d4e47b450cac714aa1455 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6830b1111a18469fb22a53769828c9f2 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c53cf6ae70f43d7bfbf19ebd3d2dd0d : FAILURE in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d600efd66347d3baadda37eed1e227 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1bd96d7e58d4ceb88c6fb0af258b8a2 : SUCCESS in 6m 12s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/83303185d2d343928399a72771a9d7a1 : SUCCESS in 8m 24s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c82ba90312ea44b48e95e41f1751b046 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e71f75687066d79cae9aa04af00f0f7b5310b26f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 23:49:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83f9014ea6eb4b73bc48e148b9e19279\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8e9de5b6f0d42e7b84157b32d34dd94 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66f3d6f86224416187a45c3808c59c46 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3798120bac1d406390a2b14369140cb1 : TIMED_OUT in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24cdd7e0f3d7f74407f240e09ed4f02346a81800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-06-29 02:03:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2c610bd80f8b9ab388edeb92987793838dcae36a","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-06-29 02:16:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"09a56af83f943ebc46b414db2954fc853c32c41e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 02:27:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768ef56849cd4179a1c08454f33e9499\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f856f74177ef40bfa368f736a2134964 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5016f494dc2a449587c02d880cc4528a : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2f0adeed33f40ffb786e68d2655494d : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1cb095d87f49ad92c1e1747302b3c8618e05c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 02:35:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b643c5a4d0e4c6f868141548e322973\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebcb30bc1f144df598964ac8c3016922 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b0e77ff8610476c9fcea281139e8672 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bbf1564476344f89f00e5484bd662ae : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11f2d8fae2844c599a72a1c24e8bbd8f : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cbe259c170a413a99ba0b9c8df99401 : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ce1731220c44708bf1eb55d6420e0b8 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf01e23740944ba942d0ef33b0b8e08 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dd956ce9b6443bc828643f68917daa5 : SUCCESS in 6m 15s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/9f343f1919f54d16a77507398432727d : SUCCESS in 9m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/fd7ed80a324a4fce9a04707937a2bb0d : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"455b87ee55436cc215929bb7d1fa8d3955619b9f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-06-29 05:42:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"095edd01f2ba1b93548105c0330c44ad4c255d5d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-06-29 07:23:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cce5a9b880f916cd839c693ed91feaa801c74956","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-06-29 15:52:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a5fdf8872bf5930361179b5bb2a59db54db31b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 15:53:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1c37ed52876bfddaa19da902bbb2db5ff9094ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:12:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f3129328af45d08fef6f22000d6a2f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/823900cb2acb4a94867c00c16c7330c7 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb82097fc76f42329e55078bf709b2cf : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8320478a60c4eefaf36e775e9ad330c : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ded6a0bb21c4aafa673596c6e3f2b0f : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4190988da734c90b654921f0dc86511 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c7c7c5f505466fac6ffe70304b6c71 : SUCCESS in 17m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a9554e5495546768c7c8da3fd58e4b9 : SUCCESS in 7m 16s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/347685cef4354e4c8c9d4db72ffee3ba : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ac9853755ead4b8c29517bb74103bdc8adcc3947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:12:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"95d9bf82b390359b111aece97ab7b22f380ab0c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:14:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/484e00518d784e6ea980ed3e5dcce188\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2165d9f5f7f741938032057a2c206fb0 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a17c008ffd4ce989ea7d00e73bb090 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f2636e00ebe445db9173c804dd0edb2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a600895ba3fd392f7e3cf68752f6e810f3afd3a0","revisions":{"1331637934fcc4f58210ffe95da181807cc45fca":{"kind":"REWORK","_number":1,"created":"2023-06-28 13:30:19.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/66/887166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/66/887166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/66/887166/1"}}},"commit":{"parents":[{"commit":"76a920aed2b103d72552c38b579769dced8d2e22","subject":"Merge \"Handle SAWarning around allocations FK Constratins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/76a920aed2b103d72552c38b579769dced8d2e22"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-28 13:29:53.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-28 13:29:53.000000000","tz":-420},"subject":"DNM: Test class based test distribution","message":"DNM: Test class based test distribution\n\nChange-Id: I5a65accfb7e2033690b2934d874141db7f4bf383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1331637934fcc4f58210ffe95da181807cc45fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/1331637934fcc4f58210ffe95da181807cc45fca"}]},"branch":"refs/heads/master"},"50291ffb19c0998543a0f89a4b45921c6b774b98":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-06-28 22:45:59.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/66/887166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/66/887166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/66/887166/2"}}},"commit":{"parents":[{"commit":"bb75e243dc39906387eb241b1e1af5f96598e19d","subject":"Drop databases in unit tests instead of contents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/bb75e243dc39906387eb241b1e1af5f96598e19d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-28 13:29:53.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-28 22:43:22.000000000","tz":-420},"subject":"Execute tests by class, not randomly","message":"Execute tests by class, not randomly\n\nThe default test runner behavior is to distribute various classes\nacross many runners, which is fine. But when you have lots of setup\nto ensure an environment is correct for tests in a class, you\ngenerally want to execute that together instead of separately.\n\nAs such, set the --parallel-class execution option for stestr.\n\nChange-Id: I5a65accfb7e2033690b2934d874141db7f4bf383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/50291ffb19c0998543a0f89a4b45921c6b774b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/50291ffb19c0998543a0f89a4b45921c6b774b98"}]},"branch":"refs/heads/master"},"a600895ba3fd392f7e3cf68752f6e810f3afd3a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-29 02:03:55.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/66/887166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/66/887166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/66/887166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/66/887166/3"}}},"commit":{"parents":[{"commit":"c392814ca8efd735d716bb900f77ec91896eccc5","subject":"CI: Fix PXE Ananconda cleanup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c392814ca8efd735d716bb900f77ec91896eccc5"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-28 13:29:53.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-06-29 02:03:55.000000000","tz":0},"subject":"Execute tests by class, not randomly","message":"Execute tests by class, not randomly\n\nThe default test runner behavior is to distribute various classes\nacross many runners, which is fine. But when you have lots of setup\nto ensure an environment is correct for tests in a class, you\ngenerally want to execute that together instead of separately.\n\nAs such, set the --parallel-class execution option for stestr.\n\nChange-Id: I5a65accfb7e2033690b2934d874141db7f4bf383\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a600895ba3fd392f7e3cf68752f6e810f3afd3a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a600895ba3fd392f7e3cf68752f6e810f3afd3a0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
