)]}'
{"id":"openstack%2Fironic~888654","triplet_id":"openstack%2Fironic~stable%2F2023.1~Ic8d3cbcb0c25683e7459362a9d81ed532607d5bc","project":"openstack/ironic","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-07-20 16:10:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic8d3cbcb0c25683e7459362a9d81ed532607d5bc","subject":"CI: Handle unhappiness with github cert","status":"ABANDONED","created":"2023-07-19 22:18:56.000000000","updated":"2023-07-20 16:10:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":888951,"cherry_pick_of_patch_set":1,"meta_rev_id":"01dafa2e036795b6613b7daf2f8d4f0a4c7d6e58","_number":888654,"virtual_id_number":888654,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-20 00:10:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 22:18:56.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-07-19 22:20:36.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-07-19 22:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 00:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"878f29049434b740fe8df4c0a374fd92ee88e0f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-07-19 22:18:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3e3f9acea483a9e4efb0ebc010d710c3b2ff78ff","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-07-19 22:20:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"73ff9daa31472dd8b867701c89b72f8e61fab566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 22:25:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1d862eaf55407b8e7099dd143b7be6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f22a586f00a426cbdc56dffdeb7c611 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4c9f4841509446785e71f314c9771a4 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89a80cde4d924eb54c6303b5e65c50815eb00f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 00:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69989968f5a344bd87529f5a8d9cba46\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3055b3eefb3f417cacb004f137775f77 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5358675c4b437e9129bb0a44cbe7c3 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aded8c7c015845f08d92e18638044f3a : SUCCESS in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54907ec9d8f24b21b44fe5e7e7b34fee : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a600c3e54c6a44b08df9f43fbb952732 : SUCCESS in 11m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/02f18365cc434e3dbc7247ffbc51b84c : SUCCESS in 7m 35s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/c71b67340407400f9d613842c43a5a0b : SUCCESS in 20m 02s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e611af4121c14cbbb21e610657e8d4df : SUCCESS in 22m 07s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/be37cbf717f64be1ab3dc4d625761d4f : SUCCESS in 48m 09s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/efe690098c7f41bd9e0d8575aa64d5ab : SUCCESS in 43m 50s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b729c8e5e6534eb19b1c1ea37c357741 : SUCCESS in 46m 04s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5e551272074a432984f0de2d272979c7 : SUCCESS in 31m 53s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ddd2bb2d5e4a475082ea99e1c3641e0e : SUCCESS in 47m 12s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/2bf14a78d83049029f19b1136112dafe : SUCCESS in 34m 55s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/30bb7b10d4a84d578e1cf8f1d534d2e9 : SUCCESS in 43m 52s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8bc1bade12184b58ac95c4eeb290f056 : SUCCESS in 1h 49m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/95a8dfae5e4a4120951d776b26e3a833 : SUCCESS in 49m 09s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/2712b4bcc9d848b0ad1ed5d565d1a175 : SUCCESS in 29m 02s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/961047e6868b46a8b11b98b9d3a2cd1f : SUCCESS in 34m 47s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/733ae9cd02b54cd081af3a8cef4e73af : SUCCESS in 1h 02m 00s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/b68f26abef1f41078ac2cd0e602e0b31 : FAILURE in 4m 14s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/09a5b6b0cae845f4b1134a1146c5c76e : SUCCESS in 47m 13s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/264911c1e6f142779ec59b807a453212 : FAILURE in 44m 31s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f51771a8c23941cb93c18c558d384374 : SUCCESS in 52m 19s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/48090d61114c4e2cbc3a46cbc8b8f6aa : FAILURE in 59m 27s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/55cf328d8e1c4fc586536cc3fa1312e8 : SUCCESS in 47m 13s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/b968a9d326c2488b8467fe20fe283466 : SUCCESS in 1h 04m 24s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da1520f14cc8485bba98959d3f28c1ef : SUCCESS in 34m 16s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/fa0e297c468049e89629aa19c40b1f80 : SUCCESS in 35m 35s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/45a8af2547f34d50832411ab379fa4e3 : FAILURE in 39m 50s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/ac2b6968a63a4eac8405cf34164ebfe8 : FAILURE in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01dafa2e036795b6613b7daf2f8d4f0a4c7d6e58","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-07-20 16:10:27.000000000","message":"Abandoned\n\nReal bug is ubuntu #2028170","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12658335cf7f74e5905510f8ec06b3cff3b937ca","revisions":{"12658335cf7f74e5905510f8ec06b3cff3b937ca":{"kind":"REWORK","_number":1,"created":"2023-07-19 22:18:56.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/54/888654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/54/888654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/888654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/888654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/54/888654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/54/888654/1"}}},"commit":{"parents":[{"commit":"7bb46729bdb30e3dce9e03e0a7db9952ec270874","subject":"Merge \"Fix self_owned_node policy check\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/7bb46729bdb30e3dce9e03e0a7db9952ec270874"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-07-19 19:22:47.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-07-19 22:18:56.000000000","tz":0},"subject":"CI: Handle unhappiness with github cert","message":"CI: Handle unhappiness with github cert\n\nChange-Id: Ic8d3cbcb0c25683e7459362a9d81ed532607d5bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/12658335cf7f74e5905510f8ec06b3cff3b937ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/12658335cf7f74e5905510f8ec06b3cff3b937ca"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
