)]}'
{"id":"openstack%2Fironic~893151","triplet_id":"openstack%2Fironic~master~Ic90f7c93d9a7b86fbf3f2cdef46bc1b2bbea489d","project":"openstack/ironic","branch":"master","topic":"fix-devstack-ipv6-ping-condition","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2023-08-30 13:20:23.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-08-30 13:57:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic90f7c93d9a7b86fbf3f2cdef46bc1b2bbea489d","subject":"devstack - fix IPv6 ping","status":"MERGED","created":"2023-08-30 10:48:44.000000000","updated":"2023-08-30 13:59:08.000000000","submitted":"2023-08-30 13:57:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893151-fix-devstack-ipv6-ping-condition","meta_rev_id":"43fcd53a72e7ceeaa432a481a8b475acd02bd5f6","_number":893151,"virtual_id_number":893151,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 13:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-08-30 11:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-08-30 11:00:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 11:00:09.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-08-30 11:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-30 12:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 13:20:23.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"8eb17204af8b97d034ff1898d0ead0287db794c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-08-30 10:48:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24d7a29f5ba94c825c2c1f23c102353ac015d28f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-08-30 11:00:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81988174561e13f758e16692997987e758fa380e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-08-30 11:00:38.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01e21cf521654c4fcf25d2d21aed6417dfc511ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:10:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fc3f281cb844beea7b19a162b0eec8a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/724d3210600348978ed698ca9db06553 : FAILURE in 16m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ac596a029f84e38994538be93c8d70d : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e12d953489bf4b7b806bc16e3ef6ea3e : SUCCESS in 17m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b77d147ce834a1f9d6611a4fa9369be : RETRY_LIMIT in 1m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"999bcac788354b48a4f3faf5c1df74e9aa4b9bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:27:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e5c5b7c28e4efba66813951ebaf408\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcd0c89fcba5445786bad3cbf1831ff4 : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d21e5d1c2f4340bec0f4dda4946563 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc11721eae2f40cf888d9b9ee5debfef : SUCCESS in 11m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d09ddeef5b56412d93649d9068e60622 : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c21da7df612146afa4905b2d0407c495 : SUCCESS in 8m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e322a204e3b4f6fb58898421267bbe5 : RETRY_LIMIT in 2m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a58678423547499a82580fed2ccde8 : SUCCESS in 14m 04s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/6742d7d2ad094bd3a3c993a2bf17084b : SUCCESS in 7m 11s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dd0bce4987d44c43809c1a8f8e7e8e3f : SUCCESS in 9m 17s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/305b115bb185480a9dfc41750a16dc51 : SUCCESS in 10m 17s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/75eb1525b5574a108755fa4942fbff44 : SUCCESS in 35m 10s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0008d66f30fb47d191bd4b84a5058f7d : SUCCESS in 29m 59s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/48048b32643e4967af149257737c54d2 : SUCCESS in 1h 05m 54s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a670173fe8f0413ca3df6481c84790ed : SUCCESS in 57m 01s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/574c9173159a4bb1838598bf359e87f9 : SUCCESS in 1h 06m 18s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/010969ed7fa1429cb12f270ecf64e096 : SUCCESS in 1h 14m 23s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4f0642b8dcb242f2a8da45ddc9bb827c : SUCCESS in 42m 10s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b81b4e32ac864dd6bcd7c5fdd4690518 : SUCCESS in 46m 35s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/71c5f4b04c074cbb9c9f1c4323e7f17a : SUCCESS in 42m 17s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/0fb3c6a12e9a442aaae6cf6c03a70102 : SUCCESS in 59m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/e62c4d2744cc40b99e9c2df30f8c8bda : SUCCESS in 1h 30m 31s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/989ca63a19d44759914ba2bb1c37ee0d : SUCCESS in 50m 33s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/0d23efef05374d0090a5e4158b7ff58e : SUCCESS in 32m 12s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/4cf4b713accf464990822282078e7a2f : SUCCESS in 52m 26s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/ffd37a4b121746568fa24b3dac4c0d3f : SUCCESS in 1h 28m 54s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/d6f345fb9d0d41dca8b759a2f754ba03 : SUCCESS in 36m 10s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/e5a1675e302d4a999cf74b2765c8a97e : SUCCESS in 5m 23s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/06c90df8787d430d930d6610d5b7ac83 : SUCCESS in 54m 38s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/7295b0b1e18f4432addba5a3f0268293 : FAILURE in 1h 03m 43s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f84c3d1b1bb54a818f235528fde5c0c1 : SUCCESS in 33m 49s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/4ee4c3217b1d4ac9801b1d72425c147a : SUCCESS in 54m 21s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/758fcf80777d4adf9100449d8d31b5f4 : SUCCESS in 45m 52s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/bf48663e013d47c28d410badb9ac3ef8 : SUCCESS in 1h 05m 13s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ece6b7e9c5d4484bc0070ceaf73be35 : SUCCESS in 39m 50s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d2eb9e4bea254dde9139ffaa96ced114 : SUCCESS in 49m 39s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/05aa11f694c94afdac016ae601428fb2 : SUCCESS in 1h 01m 02s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/a72272d9ae674055b927ad4a14f6f13d : SUCCESS in 43m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7edb0386c16e3781934476268147c175f2d1120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:28:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f74ec9430569bfc3b6d428ea71ee1555113492b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:57:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d79ef53ef6af4f14840b6f7d74e252e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4fb323077734423a66a640c83a7ff15 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32999f4ae2624c08a6f4bea1594f94e2 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dd108912ded473babe2a576bd8e3520 : SUCCESS in 8m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0d8aa97a9e14182b303d1890f57c04d : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab66562b53144f783ad63a529e0a3b9 : SUCCESS in 16m 15s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/fe45e7791e9a46e9b852c385f027aabf : SUCCESS in 5m 24s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fe2d1a80b0434a2f91517e06686b6c83 : SUCCESS in 10m 18s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/57b77a0a9bdc4e9e8c327613318f6576 : SUCCESS in 9m 11s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/96ab09dc325641ee9be7d188fa9938d4 : SUCCESS in 34m 42s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/2d5c145a6f0e445e94c69a8d5b73c72b : SUCCESS in 57m 16s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/427275c5d82145a38925169523ba4e98 : SUCCESS in 55m 58s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/86384669b5054b87984836e8388ebacb : SUCCESS in 56m 38s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2fc3c084099e4caeae046a7b7e5b3efd : SUCCESS in 34m 17s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1e288e74c73f4fd5a1f9ea5ce53d1f7b : SUCCESS in 58m 33s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/7908a71cf03847f1896f65cec942e43e : SUCCESS in 57m 16s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/095673e768304ab1828a4127d086fbce : SUCCESS in 55m 07s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/36525d1aaeae43c2817dd67c1a2d6f8e : SUCCESS in 53m 06s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f2ffe3abab7143f4909947be8ad384b3 : SUCCESS in 45m 00s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7868a6902d6b441e8cabc4502a539e27 : SUCCESS in 1h 00m 28s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/88dc7d57e04742919ddd505a2832db9f : SUCCESS in 1h 24m 00s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/451d717ba7f14a6f9e83f284d9e55cb6 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"009fdf5547b689e0cf363a9477856830e74a4948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:57:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"43fcd53a72e7ceeaa432a481a8b475acd02bd5f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbadc19fde61405eb74395e0d9025219\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b752a7ed6e8a47f69569be96fe83490d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa2dad9f7594682e3118eb9d940f883a02cdd553","revisions":{"aa2dad9f7594682e3118eb9d940f883a02cdd553":{"kind":"REWORK","_number":1,"created":"2023-08-30 10:48:44.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/51/893151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/51/893151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/893151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/893151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/51/893151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/51/893151/1"}}},"commit":{"parents":[{"commit":"c56440cc2f5177bdfbe83063463f4236afe89084","subject":"Merge \"Revert \"Move metal3 integration job to non-voting temporarily\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/c56440cc2f5177bdfbe83063463f4236afe89084"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-08-30 10:45:44.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-08-30 10:45:44.000000000","tz":120},"subject":"devstack - fix IPv6 ping","message":"devstack - fix IPv6 ping\n\nRemove the $ in the condition so that we don\u0027t attept to\nexecute the output from ping (i.e PING - unknown command)\n\nChange-Id: Ic90f7c93d9a7b86fbf3f2cdef46bc1b2bbea489d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/aa2dad9f7594682e3118eb9d940f883a02cdd553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/aa2dad9f7594682e3118eb9d940f883a02cdd553"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
