)]}'
{"id":"openstack%2Fironic~893379","triplet_id":"openstack%2Fironic~master~I3c76ddfb89b92d3d4bc29b7ccab4362604775568","project":"openstack/ironic","branch":"master","topic":"ovn-devstack","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-09-01 05:36:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c76ddfb89b92d3d4bc29b7ccab4362604775568","subject":"log the version of the conductor starting","status":"MERGED","created":"2023-08-31 14:22:11.000000000","updated":"2023-09-01 05:38:13.000000000","submitted":"2023-09-01 05:36:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"893379-ovn-devstack","meta_rev_id":"5aadbbb39835d0eeb47e610562ec589c24dddf95","_number":893379,"virtual_id_number":893379,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 05:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-08-31 14:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-09-01 04:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2023-09-01 04:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 14:23:14.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-08-31 14:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-31 16:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-01 04:08:45.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"43e13d9d0d1688d73113bd8b04fd3df173852994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-08-31 14:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b04b25fa9739f7dd0d21f20e97868a3af782f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-08-31 14:22:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fe2fd30daa9b1a3d00eba38f904fcb3f93da3700","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-31 14:23:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"628b9ad1fb601faf6e8661c766dbb3d3e5edb9b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 14:31:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe543d569d95490bb7303ab062838107\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8da9e6eec2c44d0c91d48cbc8f03eaf4 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e1cba30e1994c7589c4775b547ad957 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/260197555a3c47adbb46930faaade74d : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5fa1019b244741728e417001b76875c2 : RETRY_LIMIT in 1m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"031a4611e1bfc0c1115362c50ebd23a2a1e35336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 16:11:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/434011e8615b49f5a02456c5a9ba82e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d727c2b9f084370bd40e4216eed0aec : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c2644190ec46549677bfc0ccb3df56 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4348f8839709403cb99f12514b911e6b : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/407e7da4e93e4745aae531e72e313f06 : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c349395d5ea24aac905016900ba7f42d : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8c6a4d26a1f4d9aaf6fbb7aafae73da : RETRY_LIMIT in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b2fd6ae46241b9b083169359f6eac0 : SUCCESS in 13m 52s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/dfd6fa492bfe4090b2cff40a4e360985 : SUCCESS in 5m 42s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/64dd2bfe89a1416d8eab449e7a21241e : SUCCESS in 10m 42s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d0c9b83d4e8146ed8960ac3d9876290f : SUCCESS in 9m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4d4dd66b1eaa4169893667b545bfbe64 : SUCCESS in 9m 19s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/09a75800579b40399fc6960c37460870 : SUCCESS in 23m 59s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/07788635e6b44a39b751ba432d579d40 : SUCCESS in 26m 06s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/4133918fac4e4113ac42a9365c169db8 : SUCCESS in 58m 42s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/df9387fae47d40678bbee28b6be9acbe : SUCCESS in 1h 01m 38s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/d9b281bfe81b415caae93874c7d1c467 : SUCCESS in 52m 51s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/66f35edeb7e8447f8d86d2fdba2a0ca3 : SUCCESS in 52m 11s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7a0e4002bec44cb497b59935a2d20db4 : SUCCESS in 46m 43s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6d96c59377a94dae8ad87d31484cca41 : SUCCESS in 51m 54s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e0ef5c6dbbbf44c091b094e38090194d : SUCCESS in 49m 52s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/d71ddfe568d24b309ab877c189e6cd0d : SUCCESS in 54m 21s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/547c243ed5774c37963f28f9ff748460 : SUCCESS in 1h 42m 42s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/fc21a846f666483c8f332b8f279ed9b9 : SUCCESS in 1h 05m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/c21d043a55184fa29e844e6fd5c8ffd8 : SUCCESS in 54m 16s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/7f99f9527ac749e5ac900a6626e3923a : SUCCESS in 56m 47s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/981ebc155a0847569e9eb2ca0b957d58 : SUCCESS in 1h 19m 02s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/fbaecac1ecda483b9ee05c0ffc5acabd : SUCCESS in 40m 59s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5b03e14cdef842e5b10c9d30f30dd4ed : SUCCESS in 5m 04s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/de93deb59e7b4c04a1874c89d61ae309 : SUCCESS in 52m 49s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/d72fe90632284d87bc607e77fe5704e3 : FAILURE in 47m 21s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/4116bbc82e6e455fa8bc67ba5c841bf5 : SUCCESS in 52m 54s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/3787e60b216f499d8fe8705c27a6062f : SUCCESS in 1h 03m 39s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/4690f3c08eea4b7999f8040f2c7d5a21 : SUCCESS in 52m 28s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5e8e275dc3d74a4999c91c0c891f3467 : SUCCESS in 1h 01m 12s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecfa126f923347309961a3f7e11b8e0e : SUCCESS in 40m 23s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/e98027c1d6124a52b6f1614264a9cc85 : SUCCESS in 34m 03s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/cb114e4b002244ce835ec9282eefaf37 : SUCCESS in 54m 31s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/f08bb9a2fb6f404f86465960adb3c76b : FAILURE in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6d69ad3c9b8bedfb71610a4206a52acc04a5b7c","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-09-01 04:08:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a780fc1d798cb41e1d94f08597b65f02a3f3fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:09:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1900d5492d41cd9a3f2684ee48c2ce9cb312b5d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 05:36:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67b2e4dd0ef4fe5bf8cd5f1675ec2b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ac4779af8a4d3dabb0439d79acaebb : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/713f327a65de4062abb9c5dbade15b9c : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd46ddc68c64409fbccf7b05da1ad93b : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3b4454e9625475393648803f1d5513e : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec78f0c8f7b440686f71616e078d755 : SUCCESS in 13m 36s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/d7d89b8bfbf04728afe8580262f47e77 : SUCCESS in 5m 52s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ea3b8968ca504ef3a19f3afb8b539705 : SUCCESS in 7m 40s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d88e264399d14ec8b7a761329cdffbce : SUCCESS in 8m 44s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/606b882113114bcaad68950d8f7c7859 : SUCCESS in 21m 40s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/1d854f5168144d1188a707190cd4aec8 : SUCCESS in 43m 58s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/3920bf3431f14b14be4327f99713d2f0 : SUCCESS in 1h 03m 10s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/ffdea493b88340bf907e42c605c802b6 : SUCCESS in 1h 01m 46s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c4d746706bb74bc9817b05125e63851e : SUCCESS in 29m 29s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/48d0b9049b474dbfa00047ae7e61c21f : SUCCESS in 59m 46s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/5c7a3b964afb46e6983bd4ebd47fe63a : SUCCESS in 29m 09s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3c1f52927c3c432487d6f8925fa234ad : SUCCESS in 39m 14s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c82b57d4bca94949a21393e38a4dba7b : SUCCESS in 1h 21m 39s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/589266b9325f485db8ed3df3b8f31ed0 : SUCCESS in 42m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/88d3ca61b52c465b8bc2e8e1d4563718 : SUCCESS in 56m 08s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/0489b3c4950c4cb5b54f3c47f3a20007 : SUCCESS in 1h 22m 42s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/77254798e79342189bceb84b74601a94 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ec0eee5f1756967028c435aed246a096c426bd30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 05:36:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5aadbbb39835d0eeb47e610562ec589c24dddf95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 05:38:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8862132d64eb464fbe63a81ece3b3164\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a620ca519f1b4cb094127379af24bd7c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d2243b916ef5503f3524a7424ca69a963dc2d16","revisions":{"eafd8167b10c54d08d1673b789ec223947696574":{"kind":"REWORK","_number":1,"created":"2023-08-31 14:22:11.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/79/893379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/79/893379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/79/893379/1"}}},"commit":{"parents":[{"commit":"f9dda8def7248566799d87507e0b83ac5dce3aeb","subject":"DNM: Test grub+ovn-dhcp-opt-210","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f9dda8def7248566799d87507e0b83ac5dce3aeb"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-08-31 14:20:47.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-08-31 14:20:47.000000000","tz":-420},"subject":"log the version of the conductor starting","message":"log the version of the conductor starting\n\nJulia, who has been looking at Metal3 container logs a lot, is tired of\nnot being able to easily figure out when on the timeline of the git repo\nthat the container\u0027s source code was running from. So instead of guessing\nand having to figure out based upon behavior/logs, just log the version\nthe software believes itself to be using the existing version info,\nsimilar to how we do it with IPA.\n\nChange-Id: I3c76ddfb89b92d3d4bc29b7ccab4362604775568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eafd8167b10c54d08d1673b789ec223947696574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/eafd8167b10c54d08d1673b789ec223947696574"}]},"branch":"refs/heads/master"},"6d2243b916ef5503f3524a7424ca69a963dc2d16":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-31 14:22:23.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/79/893379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/79/893379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/79/893379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/79/893379/2"}}},"commit":{"parents":[{"commit":"a335e42b32fa5c4f624c88ec1aefae421921df86","subject":"Merge \"Add missing release mappings for 22.0 and 22.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a335e42b32fa5c4f624c88ec1aefae421921df86"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-08-31 14:20:47.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-08-31 14:22:23.000000000","tz":0},"subject":"log the version of the conductor starting","message":"log the version of the conductor starting\n\nJulia, who has been looking at Metal3 container logs a lot, is tired of\nnot being able to easily figure out when on the timeline of the git repo\nthat the container\u0027s source code was running from. So instead of guessing\nand having to figure out based upon behavior/logs, just log the version\nthe software believes itself to be using the existing version info,\nsimilar to how we do it with IPA.\n\nChange-Id: I3c76ddfb89b92d3d4bc29b7ccab4362604775568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6d2243b916ef5503f3524a7424ca69a963dc2d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6d2243b916ef5503f3524a7424ca69a963dc2d16"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
