)]}'
{"id":"openstack%2Fironic~897071","triplet_id":"openstack%2Fironic~master~I1aeeda89a8925fbbc2dae752742f0be4bc23bee0","project":"openstack/ironic","branch":"master","topic":"workers","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-10-05 15:02:15.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I1aeeda89a8925fbbc2dae752742f0be4bc23bee0","subject":"Bump workers_pool_size to 300 and remove queueing of tasks","status":"MERGED","created":"2023-10-02 16:44:47.000000000","updated":"2023-10-10 07:19:00.000000000","submitted":"2023-10-05 15:02:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897071-workers","meta_rev_id":"648b63d9a6062079f6c760911f807a9acbaeae80","_number":897071,"virtual_id_number":897071,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-05 15:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-10-05 13:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-05 11:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-10-05 13:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-05 11:54:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-02 17:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-02 17:27:18.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-10-02 18:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 11:54:50.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2023-10-05 13:27:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"d5f63c2415dbe847e43006fa2769621ad1d6a91f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-02 16:44:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81b43ecbe9309d4b3544e691c02f5422f2b20f8f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-02 16:47:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8cf78daab36e39c6d7c3c34c393db4bd723d5a22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 17:24:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8376b8099f9e4418b3956cdbcddf8489\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd89c9cbb3044095a5692f842e670452 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76f7a30e7ce043cc9854343005a929aa : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb46f65a2389476c8dffe5b3e2ba6d88 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d2be12317134ad5921f8f122a233b1f : SUCCESS in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11997538623339d4a77a97951a355a8543ac510a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-10-02 17:27:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/7dc6db4/7dc6db4.html : SUCCESS in 41m 18s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/48d79b7/48d79b7.html : SUCCESS in 14m 01s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/7f46689/7f46689.html : SUCCESS in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85eb0f652b83a0c39761c69d1692b4137b93a9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beacfc67b32d4adb9f469b673c402d1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9155b9d267a94cb2b69974689d58c6d0 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92abdad88f64dac8daf15af2197721f : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5bb7beab02942a09cc3c508d4eafec7 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c42d087d03c2420d9ededec5ad5d38d1 : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f576d78e2e84189867c352b0a29fa3d : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/949bc0d31af1424f846bc3b300cd055d : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d75157d80d42eda96b828b16a51ec6 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e8091dae8964005a81590f9055e3818 : SUCCESS in 7m 41s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/09bb7a995b4149768b9e15ffea81dd9e : SUCCESS in 5m 08s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a443a21c41c04614b300ecd12b155314 : SUCCESS in 11m 05s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c4afee8fa1894c08bfbd21d9a489c456 : SUCCESS in 8m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/0097084877e64b0d908516227a639154 : SUCCESS in 9m 08s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/1ccf1bd846994d73be098fe84a44c9ca : SUCCESS in 24m 21s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/1b32c1862181451d86de9d79114b7012 : SUCCESS in 32m 39s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5550983964d14dae9586922668095e5b : SUCCESS in 1h 04m 17s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/594a8ea21ef147d2a981d7d44d76f59e : SUCCESS in 59m 39s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/0fbeffe5c55840bc883d78363af87f24 : SUCCESS in 46m 36s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/664d8fb7db2647599a936a94b476b44b : SUCCESS in 37m 39s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4cf59eb67ab5417d8d116d937959fccf : SUCCESS in 44m 48s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/46ec96971fcb4f14bfeace87a37f0d1f : SUCCESS in 57m 55s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/1fab295759424c43981866185b46c9e4 : FAILURE in 14m 17s (non-voting)\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8b42043e8d904cb6be4dcbfc8e175c54 : SUCCESS in 55m 48s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f9babaeeb9524d9d886d0d612996b59a : SUCCESS in 1h 14m 36s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e5c0fb7b6bbb43c5a41b57e6832f4ee5 : SUCCESS in 56m 27s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e9d695c6e8f842d999a9df2250ae2bf9 : SUCCESS in 39m 31s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/c537514a206a4aff88f0f2eec1fd03e2 : SUCCESS in 1h 00m 56s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/950a40f921b24ef181c0106fedc456bb : SUCCESS in 1h 30m 50s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/9a41d387d1434b6ea83b7ff07e2c28f8 : SUCCESS in 40m 26s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/cf1317a04f3d4d46aec092327f594ddd : SUCCESS in 5m 07s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/15a3d8776764452d845f31c9d43b12dc : SUCCESS in 1h 02m 58s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/72b8009bdc2a41388ce53e96144992f4 : FAILURE in 57m 08s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/513c099ff5db4425948c0401afa20422 : SUCCESS in 45m 38s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/70285a786f194804b7b29bdfd3a8b2f2 : FAILURE in 1h 00m 51s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d3a3312a250d459c8f4e9a98b0fecdae : SUCCESS in 40m 38s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4059c7a6747e496f99f8ab39ab95ba34 : SUCCESS in 1h 03m 48s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52945cbe61e7435a8286f737b461605d : SUCCESS in 40m 43s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/cde971aa707c46e494534da11e1e511c : SUCCESS in 34m 31s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/fedf625ca0064cc48f1a51a4004aa107 : SUCCESS in 51m 39s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/0a99c1a08b074cb3a42c28dfc8b434a9 : SUCCESS in 42m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c649ca02c1399f733d895c7606caed9fa6210d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-05 06:52:16.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9896788b6f082c32c62ae27c0405dc439cffffb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 07:11:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa3dde4d1654150af117fb36415d39c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6181a890af8142779132694a5c34712a : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f06c52dea29420f90a2645abb5e6017 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a5bdc64ddd34832aaa4161c5dad928e : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a28dbe8457c43d5851e9eb453038f3f : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc26e2578ea7373d4250d238acbaaa1e9fc50638","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-10-05 07:35:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/b3d9467/b3d9467.html : SUCCESS in 41m 41s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/9aaaa61/9aaaa61.html : SUCCESS in 15m 04s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/71/897071/2/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/544b47a/544b47a.html : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c36a3267984f2ef449ff260f10a929f2c49967f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 08:47:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0d589618205424a917c71dfd35b0424\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53502faf889421aab56736db53d0a98 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfcc4cc487f4003bfd42b8707dea69f : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d556f96b76340d1bc953165510d02ba : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843a933664af4d9484711a16c237da4d : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f8b1df327e14224b3799c73705f0f4c : SUCCESS in 8m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d3a41542b68492b99ec95b6ccc80db3 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a6235ca5ef44cc9702a4ade90e2934 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265e0536633e4979830eb833252a409b : SUCCESS in 7m 35s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/98e773d0ed9b43b59a51867630ce60c4 : SUCCESS in 5m 20s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d95e70a4e4414ce78664d8b476d991dd : SUCCESS in 10m 19s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/512413ce6f81425097fa7bcb2ae15899 : SUCCESS in 7m 13s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8d56a48412214c7d8443becd2e9058e1 : SUCCESS in 8m 51s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/5c57436a1a9a442dbdf9bf8ad9d35445 : SUCCESS in 30m 27s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/776720a5a5dc4967957de8b19992ea67 : SUCCESS in 33m 12s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/742443728a5348ca87477ea899f32033 : SUCCESS in 58m 35s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dea84e8525b24eb6b7466212fa4acef5 : SUCCESS in 50m 14s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/5947562c05cb48178e08820fb22a9588 : SUCCESS in 1h 01m 49s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/b24db823b454466f96b2ff5f2ac371af : SUCCESS in 51m 42s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/07408607814841429d4a413987e40c96 : SUCCESS in 44m 38s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f5aa1089393b476c892eb09258bad361 : SUCCESS in 1h 13m 39s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/d78339e19d5c425583f4030921694575 : FAILURE in 8m 46s (non-voting)\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/db185d6deab34cfea4099f2fd470f607 : SUCCESS in 55m 52s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/9ab66b9ae2d9497b8738910f88f9b5ac : SUCCESS in 1h 49m 37s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/aca7f9741f8f4d31afef01a0db9fc001 : SUCCESS in 1h 10m 29s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/6c1f782096b6481e847ce15f1405a8f9 : SUCCESS in 35m 42s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/81b97985ab7b46788cabf1419ba76e6e : SUCCESS in 59m 47s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/7a6fb28a791d473280b86549077f1673 : SUCCESS in 1h 01m 53s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/eb2a31a2cbe042bc853cbc97209096eb : SUCCESS in 41m 15s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/6dc8adcf5c144ee48304b27f8dd6c39b : SUCCESS in 5m 51s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c8cecea6cbf2410ba805577cf210ffa6 : SUCCESS in 59m 43s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/9985bb3130df4e06874432f7ab0ca5f7 : FAILURE in 50m 18s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/78ef7c29691d45198b9b1871f45db8c4 : SUCCESS in 40m 54s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/b2151d3a7acb4e7b9a239a663b910272 : FAILURE in 1h 05m 30s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/b0e96686377444db8340daa57c2165d0 : SUCCESS in 49m 32s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/beebb3a619b94e91a3de4dae231ac3ed : SUCCESS in 1h 03m 30s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ef3196852a44d75ad9d8464e6a482d1 : SUCCESS in 40m 18s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/ab8388a07a1b4505b898fda6c0f3206a : SUCCESS in 31m 28s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/37723eff20c44e9db604cda99dc06071 : SUCCESS in 46m 57s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/18ed3d0a68ce4166bcf5383dd36815b7 : SUCCESS in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b95176abefe4decc67a3b6a473415b64e1104bb6","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-10-05 11:54:50.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e25f62626849577e52eef799f847702066eb8936","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-05 13:27:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a71dd4bf67ac0ca71e0a628fe2fec69153465d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 13:28:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84842e487fc5283b13cf8f2a7d2e52be9e827633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 15:02:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c377048a0f8645a2b10eddb31b2b3eda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a973b8d19484b30aaefc9fd3728d00a : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f7894347ea4d2887d2fe3b4820331a : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65269cdad60248fa990b133498a4ce27 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfb8a54afd4747e4b1e3ee9237b462c7 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ed340751e14836840b6573fbf98ed3 : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dafd552107304a98addef2301ae2ca39 : SUCCESS in 8m 39s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1b7a5c1da0a04a9e86f88aaa390d9492 : SUCCESS in 5m 19s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e0558cb5c0734ab09b7d5f3f299a754f : SUCCESS in 11m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/33a905c39c8549f6984e16442b31d78c : SUCCESS in 9m 30s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/9f56c4afcd234813908a0b672715135e : SUCCESS in 32m 08s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/0f611c73b1c44957a12779da9637ded1 : SUCCESS in 1h 07m 25s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/29a5fc6ad53a47dcb1d797e52e7ef769 : SUCCESS in 55m 05s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/699b427719874a6384cd62f347106a8e : SUCCESS in 1h 06m 24s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9e98d654a8a24a5597c938ac9ad3745d : SUCCESS in 48m 00s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/34ac37e173ea4ef79e757992fa198fc1 : SUCCESS in 1h 01m 15s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/3569e5454dac45848e09e5e21df50bc8 : SUCCESS in 53m 17s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/bb269cd04c7343f1bbccc23cbea60fd4 : SUCCESS in 56m 58s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/d207d5299ffc4f879e4ea97b69bf7eeb : SUCCESS in 48m 57s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ecfad5f4a1c1485db79abab0b8d09504 : SUCCESS in 1h 01m 58s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/545668a79f3342cda16e1a33054dc770 : SUCCESS in 1h 28m 47s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/a8414552d863425ba8e23b72568ad6af : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":2},{"id":"374b4c257ccb5d0b62b6b1f7dfeb11f2b0e87e2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 15:02:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4b2cf9d342521353fc24b6b691ad07feaae9c7d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 15:03:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77f525ccd384efd80e069d0c78ed4c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b750cf7e430403f82cfb07b287cebcc : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5532817fa4453bbd3e6c91ec7c72da : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"648b63d9a6062079f6c760911f807a9acbaeae80","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-10 07:19:00.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"224cdd726cc0ca1826273e46619769f87380f2d7","revisions":{"2720b9ecf9464d00cc9a001b29adac2b94679d79":{"kind":"REWORK","_number":1,"created":"2023-10-02 16:44:47.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/71/897071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/71/897071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/71/897071/1"}}},"commit":{"parents":[{"commit":"db549850e0c169b11e5458722af40e87fe013ec9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/db549850e0c169b11e5458722af40e87fe013ec9"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-02 16:42:07.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-02 16:42:07.000000000","tz":120},"subject":"Bump workers_pool_size to 300 and remove queueing of tasks","message":"Bump workers_pool_size to 300 and remove queueing of tasks\n\nEspecially in a single-conductor environment, the number of threads\nshould be larger than max_concurrent_deploy, otherwise the latter cannot\nbe reached in practice or will cause issues with heartbeats.\n\nOn the other hand, this change fixes an issue with how we use futurist.\nDue to a misunderstanding, we ended up setting the workers pool size to\n100 and then also allowing 100 more requests to be queued.\n\nTo be it shortly, this change moves from 100 threads + 100 queued to\n300 threads and no queue.\n\nChange-Id: I1aeeda89a8925fbbc2dae752742f0be4bc23bee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2720b9ecf9464d00cc9a001b29adac2b94679d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2720b9ecf9464d00cc9a001b29adac2b94679d79"}]},"branch":"refs/heads/master"},"224cdd726cc0ca1826273e46619769f87380f2d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-05 06:52:16.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/71/897071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/71/897071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/71/897071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/71/897071/2"}}},"commit":{"parents":[{"commit":"db549850e0c169b11e5458722af40e87fe013ec9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/db549850e0c169b11e5458722af40e87fe013ec9"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-02 16:42:07.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-05 06:51:51.000000000","tz":120},"subject":"Bump workers_pool_size to 300 and remove queueing of tasks","message":"Bump workers_pool_size to 300 and remove queueing of tasks\n\nEspecially in a single-conductor environment, the number of threads\nshould be larger than max_concurrent_deploy, otherwise the latter cannot\nbe reached in practice or will cause issues with heartbeats.\n\nOn the other hand, this change fixes an issue with how we use futurist.\nDue to a misunderstanding, we ended up setting the workers pool size to\n100 and then also allowing 100 more requests to be queued.\n\nTo be it shortly, this change moves from 100 threads + 100 queued to\n300 threads and no queue.\n\nPartial-Bug: #2038438\nChange-Id: I1aeeda89a8925fbbc2dae752742f0be4bc23bee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/224cdd726cc0ca1826273e46619769f87380f2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/224cdd726cc0ca1826273e46619769f87380f2d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
