)]}'
{"id":"openstack%2Fironic~898237","triplet_id":"openstack%2Fironic~master~If0f185e0303d6f8071306edbc64b9c5704f58d16","project":"openstack/ironic","branch":"master","topic":"inspection-docs","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-12-13 19:02:43.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-12-13 20:11:39.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"If0f185e0303d6f8071306edbc64b9c5704f58d16","subject":"First pass at the new in-band inspection docs","status":"MERGED","created":"2023-10-13 16:07:59.000000000","updated":"2023-12-13 20:14:41.000000000","submitted":"2023-12-13 20:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"898237-inspection-docs","meta_rev_id":"b615d6b8890ed21b12d3883d97c91195f597db3b","_number":898237,"virtual_id_number":898237,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 20:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-12-13 19:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-12-07 08:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-12-13 19:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-13 16:11:56.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-10-13 16:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-01 19:35:11.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-12-07 08:14:05.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"c9e78cfaf61afabdaf0e8c8fd13322e78370bd02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-13 16:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72378f5e6c53de326cd12a87364cb4a3dbc8750e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-10-13 16:11:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/37/898237/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/ddfd9ce/ddfd9ce.html : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":1},{"id":"63d5a9852f395bcc5e63712eab855398f86392f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 16:18:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c843920fa2124eacb1ed363ce3aa500e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676754f9b5834add906f4f362f59e583 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb18efe6d2e444e8601804f258b41ba : FAILURE in 9m 38s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/1e3b9448fa6b457186112d42adbf71d0 : SUCCESS in 4m 45s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8a23525e75764a748998d1ff9783abe8 : SUCCESS in 7m 27s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/47b598b258f14d29966ad38d09560d70 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"821c398b6faf8f44490833ea9da2df279149bc0b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-11-06 16:30:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e37f073a920fc22413111f741fa3bead5e00f01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 16:48:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8763b3efe5ca41888e0e6f9cf60cec69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/436c4458a3424bdfbb11f20c9c2da5f3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2c616b9fd34c93a08df4de6fb38920 : FAILURE in 11m 44s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/aefc2fa000a14dc297c6d72fff5e8f85 : SUCCESS in 5m 39s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d3728c218b3544b793f2d718258fe20c : SUCCESS in 9m 58s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5dfd60cac55344169c66b134df7eebf0 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"92860a2e58854aa4f34477a4ed7681e8c8d03501","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-11-28 18:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5af1c4c930ce2928c8199ddb8cf7fc5e36294bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:36:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c3b1f565674e1abfc6073980862365\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ccf00dd940f400a90a655531e089b01 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113b0c6c1dd640359c43977a6fb2aea6 : FAILURE in 12m 26s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/cb7acfd8910d40e9bdac62ef355097a4 : SUCCESS in 5m 15s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/82c081b8c96349bda9508bef391ed2a5 : SUCCESS in 9m 52s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/956230f6f1b743f0abd18fdb509d7ebc : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4db9ef2243b64118ce0ff4abd777145f21b16113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-11-28 18:45:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6fa9434a76f48c73647fea119b01cd0a683a0d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-11-28 18:47:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b616e0cdc7d3c33597a294941880e7a18ac7ccc","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-11-28 18:50:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct finger://localhost.localdomain/be49e30c432b4e789b73e61e59c73264 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":3},{"id":"2e19f7f4a7f9f2fc94ec873af4056849dd48dd61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 19:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3535cdf147a84f7da49eafc46986310d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9556ddd1d98c4c78b849c01f92531289 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f048c87a78d64415beca56f4426dee7d : SUCCESS in 13m 17s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/e38ceb40b5fd44db9c008dee612e8837 : SUCCESS in 6m 18s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e2337012167e44ffb6aa7e1c58fa9245 : SUCCESS in 11m 06s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d426792da58b47818d5a03e9d1bc8697 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":3},{"id":"626dcf9af27f71017f6aca5350ce00219beefa9b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-01 19:35:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"346805062ab558064df63913941dee6ee8685659","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-12-04 15:03:32.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"cfdca73c4360c8d017088a97856efdda3ebebbfd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-12-04 17:02:50.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"be335981e2cd89e46240591014ae11179876074e","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-04 17:19:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cb534fb289e75cde08acbfcf8dba822f5694a07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-12-04 17:40:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"431e299a11c6df1bcfbe5c91dea81683b2796b13","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-12-04 17:46:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/37/898237/4/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/3dd1b47/3dd1b47.html : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":4},{"id":"5e0c934317ef2ee77449d0308805193ef4421bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 18:00:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f56ab9fe87749639efda80f9023f502\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8408da783a8e41b99654cb3c641f22fe : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36719fedad074f57975d5ce1835b4b50 : SUCCESS in 13m 38s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/7477633762d442aa8da26cdf83b9552f : SUCCESS in 4m 58s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/65f71a332330438e88c5a8036ab1bb08 : SUCCESS in 10m 11s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/0998cb2464814595a8c82efd9fdc82a5 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":4},{"id":"52be78568b3b750a40508abc4b98013ee5adaa5c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-12-07 08:14:05.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7219fc722bd6cc1637f3cba6cf3be3b0696aab6f","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-13 19:02:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bad313650f9e13c9660379a1a6c2f201d6c27488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 19:03:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7a56e42a2473609ad098a568d513527ebc036974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 20:11:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e1f739a57a046bdbca3daa2afbf1a80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d66a8914d3547d5879b13e6a16d9f76 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ddaebd5d5347029a54f644d886f99d : SUCCESS in 13m 33s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/7673fdfb703640429430e7f2a5a82c29 : SUCCESS in 5m 18s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/53bbad034e8c410db3b3e917d13b03df : SUCCESS in 9m 31s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5dbd6f0446fe4fd28d1faab7a5982691 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":4},{"id":"cc48d5d5f836e341f947d5d423304ed3bf11ac58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 20:11:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b615d6b8890ed21b12d3883d97c91195f597db3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 20:14:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d99b8cdf0e94c5d81b994efe7d241d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8efc038d64145679369cb529e728f29 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9c0996d1a3ad504420079a50451a3aefa63d5d1f","revisions":{"49950eb255cb620c8be0de3bc2446ef8536fabd3":{"kind":"REWORK","_number":1,"created":"2023-10-13 16:07:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/37/898237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/37/898237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/37/898237/1"}}},"commit":{"parents":[{"commit":"665f0617551418cd5fcabb24eacf24cde98ed84f","subject":"Merge \"Introduce conductor touch while offline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/665f0617551418cd5fcabb24eacf24cde98ed84f"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-13 16:07:17.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-13 16:07:17.000000000","tz":120},"subject":"First pass at the new in-band inspection docs","message":"First pass at the new in-band inspection docs\n\nReorganize the existing docs to give space to more information.\nCover the most critical topics, except for installation.\n\nChange-Id: If0f185e0303d6f8071306edbc64b9c5704f58d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/49950eb255cb620c8be0de3bc2446ef8536fabd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/49950eb255cb620c8be0de3bc2446ef8536fabd3"}]},"branch":"refs/heads/master"},"be482f25770494a5fa2a2bfe23c30830bf5626b0":{"kind":"REWORK","_number":2,"created":"2023-11-28 18:45:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/37/898237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/37/898237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/37/898237/2"}}},"commit":{"parents":[{"commit":"28b3f56b2c99972e044322c39fbf0965f7b441bc","subject":"Merge \"implement basic-auth support for user image download process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/28b3f56b2c99972e044322c39fbf0965f7b441bc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-13 16:07:17.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-11-28 18:44:45.000000000","tz":60},"subject":"First pass at the new in-band inspection docs","message":"First pass at the new in-band inspection docs\n\nReorganize the existing docs to give space to more information.\nCover the most critical topics, except for installation.\n\nChange-Id: If0f185e0303d6f8071306edbc64b9c5704f58d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/be482f25770494a5fa2a2bfe23c30830bf5626b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/be482f25770494a5fa2a2bfe23c30830bf5626b0"}]},"branch":"refs/heads/master"},"abd754988d2f9b562244e149930fdf6b0f92b6f5":{"kind":"REWORK","_number":3,"created":"2023-11-28 18:47:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/37/898237/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/37/898237/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/37/898237/3"}}},"commit":{"parents":[{"commit":"28b3f56b2c99972e044322c39fbf0965f7b441bc","subject":"Merge \"implement basic-auth support for user image download process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/28b3f56b2c99972e044322c39fbf0965f7b441bc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-13 16:07:17.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-11-28 18:46:56.000000000","tz":60},"subject":"First pass at the new in-band inspection docs","message":"First pass at the new in-band inspection docs\n\nReorganize the existing docs to give space to more information.\nCover the most critical topics, except for installation.\n\nChange-Id: If0f185e0303d6f8071306edbc64b9c5704f58d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/abd754988d2f9b562244e149930fdf6b0f92b6f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/abd754988d2f9b562244e149930fdf6b0f92b6f5"}]},"branch":"refs/heads/master"},"9c0996d1a3ad504420079a50451a3aefa63d5d1f":{"kind":"REWORK","_number":4,"created":"2023-12-04 17:40:58.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/37/898237/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/37/898237/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/898237/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/37/898237/4"}}},"commit":{"parents":[{"commit":"28b3f56b2c99972e044322c39fbf0965f7b441bc","subject":"Merge \"implement basic-auth support for user image download process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/28b3f56b2c99972e044322c39fbf0965f7b441bc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-10-13 16:07:17.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2023-12-04 17:40:34.000000000","tz":60},"subject":"First pass at the new in-band inspection docs","message":"First pass at the new in-band inspection docs\n\nReorganize the existing docs to give space to more information.\nCover the most critical topics, except for installation.\n\nChange-Id: If0f185e0303d6f8071306edbc64b9c5704f58d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9c0996d1a3ad504420079a50451a3aefa63d5d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/9c0996d1a3ad504420079a50451a3aefa63d5d1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
