)]}'
{"id":"openstack%2Fironic~899335","triplet_id":"openstack%2Fironic~stable%2F2023.2~I7a5e871133300fea8a77ad5bfd9a0b045c24c201","project":"openstack/ironic","branch":"stable/2023.2","topic":"handle-dvd-eject-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2023-10-31 10:50:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a5e871133300fea8a77ad5bfd9a0b045c24c201","subject":"Make sure we eject media from DVD when CD is requested","status":"MERGED","created":"2023-10-30 23:44:02.000000000","updated":"2023-10-31 10:51:42.000000000","submitted":"2023-10-31 10:50:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899335-handle-dvd-eject-stable/2023.2","cherry_pick_of_change":897989,"cherry_pick_of_patch_set":3,"meta_rev_id":"c912fb7711cbfecec8774d0f70b3141bbbcda4cf","_number":899335,"virtual_id_number":899335,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 10:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-10-31 08:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-10-31 08:04:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 23:47:55.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-10-31 00:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-31 01:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-31 08:04:27.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"5af87ff68b4790d600afda6216a8f4b479c6b77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-10-30 23:44:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"739650445336ce9938cd492545ff35e4ed51f851","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-10-30 23:47:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct https://hpproliant.github.io/hpeproliant.github.io/logs/35/899335/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct/9948d30/9948d30.html : FAILURE in 2m 08s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct https://hpproliant.github.io/hpeproliant.github.io/logs/35/899335/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct/448d907/448d907.html : FAILURE in 2m 16s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct https://hpproliant.github.io/hpeproliant.github.io/logs/35/899335/1/check/HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct/8436657/8436657.html : FAILURE in 2m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46762d2605c427a625a0fd676d500208a54221f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 00:10:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e2739f38034d31857b4b2b91b4ce1b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ca77721e2594cf28c2efa582cc4d94d : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac145958f28449e6b479c9537ac282fe : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf21890152834f78b49e3b58616d9b1f : FAILURE in 22m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/991f71b067f14ede847b4bcd440dd3eb : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4031d7cb61f5de6e1a2eb1eaa8b32ead30e023b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 01:16:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5163850add2f4bcea0646566d0ec303c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b133ec4094f46f4bf8bbc3b3987ad69 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907ef34127994f50af7ebe23b5080f91 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7ed9f984c8d41d3b03712da82e12910 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ffbb573c6e4ab69aff5b6ac7e00880 : SUCCESS in 10m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaddf3139b794879a25b34151ac3db79 : SUCCESS in 8m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/960ce87319d34bd9830f10e13006ff9c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5085765603c24a6f92d2938dd43658ce : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d958d987a14c4bcb9d87602f30d9e76c : SUCCESS in 8m 45s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/19b1f046a1ed4eee90ce942cd7d5fec0 : SUCCESS in 8m 32s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/25a0a6ab4da0493f8fee935d98036564 : SUCCESS in 9m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6490fa49829141e1b7c76d52377113d5 : SUCCESS in 8m 59s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/b91af3907c4145098c17981e7c51e4e6 : SUCCESS in 22m 51s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/587c3c583ba74f9fa6ffbf3ad23219e9 : SUCCESS in 26m 48s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/06eb5f217ab0474fa55472e873e38122 : SUCCESS in 58m 59s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c77a4d724ac7444699d588f91e355f38 : SUCCESS in 58m 07s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/040198207811407d955987447403727d : SUCCESS in 52m 26s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/18bf8d11218d4c18afc7cf5057af5e94 : SUCCESS in 50m 48s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5e24c2816f2e447898a798d41a728c54 : SUCCESS in 37m 30s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/691a68b586ca41ed981a63d14d302fe6 : SUCCESS in 49m 12s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/648cd2ddc74b40bf8e5d3f910c302e92 : FAILURE in 7m 24s (non-voting)\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/8a176da6b78b405886e7db19a0126994 : SUCCESS in 48m 26s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0d5266be40cf4d60abb87f911b03b71b : SUCCESS in 1h 19m 25s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/76f5a9070d5c4ee0b99152dd26cced72 : SUCCESS in 53m 16s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/f89e2544704643d881a2ac96dbf5f667 : SUCCESS in 33m 56s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/43edf9dcc78040a999458648850bc770 : SUCCESS in 46m 49s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/36e73b64ce5d40a8a5e26c303ecc6436 : FAILURE in 28m 51s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/802634e7d1c2438b85f6bfb3e4538a14 : SUCCESS in 38m 23s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/73e481fa420c4110872a94ca2db1e77b : SUCCESS in 6m 08s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/24783eebf4754ba48ec4b3342e305114 : SUCCESS in 49m 34s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/7062ee598a02412b87ad653a8c82c1e6 : FAILURE in 49m 07s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/b3d7836a460840ab89766dea2cd50d7d : SUCCESS in 41m 34s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/8cc6c889617f48d693d8e2164a1fb3ad : SUCCESS in 50m 26s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/70b266b10c044703aaaa70836f250e1b : SUCCESS in 35m 05s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0cd2f19cc8c74375bddd808f41198ddb : SUCCESS in 46m 12s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/109b3aabc2404d9db5bf591497bc360a : SUCCESS in 33m 40s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/43ad3e3269dd46ecbd9d60fa85ee7d45 : SUCCESS in 31m 19s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/22422c356e504b4bade0f564bc75121c : SUCCESS in 42m 10s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/607af6fde1d74ed38bb98bec581157cb : SUCCESS in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e733dd7d47a9bd13f50adf822b0ea91ac1c2b2f7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-31 08:04:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f8177d8ee0023676f5f9d47a06118558762788b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 08:16:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d7788bd3da4bb182e6089ef7cb8bdf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d686b48e8417448681d1b0719c979cb4 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7bac766aa9b2489091793e2967b7418f : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01f39a5619514c978dd99c54ece2d245 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5df8ad88326045f3b300336035488c86 : SUCCESS in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5f06500c5f912c9b50a6e78ed0eeb73cfccf346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02859bf130514a95b6fcb06b5cb21529\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7e6eb2224646b5b421ab77abd195e0 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67f95bd33d54aab82b23498170a5a49 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8286589ff2c443c6bb274aac7f47fc66 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63792c40386b44c7af9fe36c69170f5e : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fbd13fda8384511bb0d1909f8f3a38e : SUCCESS in 7m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e571e2583ac642aba317416a42f21967 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821ae366a331448f859362c9bbeec9de : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f81d6b7e2414aaf9078ca9d0810c894 : SUCCESS in 8m 23s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/38da91d7fdeb4829a8e8dfbc4b08d75d : SUCCESS in 5m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/59c38c56781d41faba102a7c48903dbc : SUCCESS in 8m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/82683cb895184a98b434842303b7406a : SUCCESS in 7m 22s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/f7d0725048fa4224a35c0c9428f461b5 : SUCCESS in 22m 43s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/0d81199d50fb434788bfacb02d10b4b9 : SUCCESS in 18m 51s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/be95f8c723d14687b81bfa7b29af72c7 : SUCCESS in 59m 46s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b6f76ffba2f4142b964e0f580cd91d5 : SUCCESS in 48m 34s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c45e65ed11c74e1bb9d762a37b729e50 : SUCCESS in 56m 20s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/0d71fee9cc9c4addb9ce6b05f9ab4c12 : SUCCESS in 47m 54s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cbc175df4c204ae399e3bc37ef02a975 : SUCCESS in 39m 07s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9401ef0f62da4de6b4b88bfc693313db : SUCCESS in 47m 12s\n- ironic-tempest-wholedisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/65fb0300f3d9461ba888804d7a81292c : FAILURE in 7m 25s (non-voting)\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/6476d4fb54e54fbc81d34273d2af2a2b : SUCCESS in 49m 31s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/fab251c279744c90bede359e1a558421 : SUCCESS in 1h 12m 10s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8e0293180440460a9646e4af9a44ffe7 : SUCCESS in 1h 00m 18s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/577de08117b345c4819792d23b445e22 : SUCCESS in 28m 41s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/d493a6338e7e416da0453f3bba60ea02 : SUCCESS in 40m 27s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/800882fcc2db4f01a8c710bd1e5174d2 : SUCCESS in 1h 04m 41s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/58881a12bfe747cb96af9fd39ad79085 : SUCCESS in 43m 17s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/dfeb1357e470455e97c6b22ee344d7d1 : SUCCESS in 5m 31s (non-voting)\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/bfa8263efc8941f9a4caec28c0ff29fb : SUCCESS in 47m 07s (non-voting)\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/ddfa4c84754140b68619e1408e7b14d2 : FAILURE in 54m 06s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/f7db369911bf42209031b1f3bd4e8b1e : SUCCESS in 37m 34s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/d5fdb8bad3214175b4f901cb2ed5bf25 : FAILURE in 1h 09m 21s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/63ccc23d63f74fe49a9d5cd0d118e686 : SUCCESS in 33m 02s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/5e5740f6a2b84e9e882764b2aa9b3d0e : SUCCESS in 45m 13s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69b9456618784a79b7caefbfdf72bc09 : SUCCESS in 35m 49s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/d55f79a027a047fe93892afe3cd394cd : SUCCESS in 32m 42s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/99d711a7440b4a1c92ea25fca6c67537 : SUCCESS in 42m 11s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/179efc326ce04cf49f6fcde429678678 : SUCCESS in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c900ced8076eec2dd28ce3d5fb73c9d5a832085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:35:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fdf8e472aaf1b61144a6223f0351c5829393715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:50:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89fd7633925b43dca836fea44dafbe64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62d17b92bb994543b1892b923ee41cb8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e60d94a386473bbb27b290314c4f66 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcca0acf56e14310bfbc0ff2568211d9 : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7c98f31c7564c19937c5b729f89d557 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488b9f5933154da297c4c8a3f4e48306 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74d939567f594bbbaaa8d91dafebc14e : SUCCESS in 8m 21s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/4b951805c38a4dcc96bfae247336905f : SUCCESS in 5m 36s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/1e49881e36ff4db6945357a6d66b88df : SUCCESS in 8m 22s\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/69b46db137c54c24ab82c3ff29af1a02 : SUCCESS in 23m 34s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/1586fa252ef54cf297376fae59275548 : SUCCESS in 57m 34s\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/a8b86f1ae3954a6cad1b606e381c61fc : SUCCESS in 55m 53s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/bedd61bfd91541c38f4215114e8a2ddc : SUCCESS in 52m 44s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/38c7cf4b855445e3be320d0f88dff181 : SUCCESS in 40m 03s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ddf69fb58e2448979a73f3d28f2f9f88 : SUCCESS in 46m 51s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/baafb5c8d25f452c81b5e234e1b0e071 : SUCCESS in 53m 03s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f04eb4d05d104806ab00e15f8e705cb9 : SUCCESS in 51m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/03f8c4b850a34831ac7ee7b42d059563 : SUCCESS in 37m 28s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/ddab56c95cd844cfb14863571158f166 : SUCCESS in 1h 01m 47s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/c9dd651dfcfc4a0b8a125984843db0e8 : SUCCESS in 1h 09m 00s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/a8a8e3173f724ac383e314090bfa817f : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"790296cd3caba921a2c8020c4a2ca03d5591a976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:50:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c912fb7711cbfecec8774d0f70b3141bbbcda4cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:51:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570b0c4c439745e48bd5f0667c1812d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ccaecbfed24f8da41c8ea89c7bc74f : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ff902b401bb42f69ea65b07eb8f292d : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"235e6ccb9835a2583984a1a47614b8bab1dbad0e","revisions":{"235e6ccb9835a2583984a1a47614b8bab1dbad0e":{"kind":"REWORK","_number":1,"created":"2023-10-30 23:44:02.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/35/899335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/35/899335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/899335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/899335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/35/899335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/35/899335/1"}}},"commit":{"parents":[{"commit":"2a523a8aa98dd4d152b7961ee17e95069096e39d","subject":"Merge \"Do not store ramdisk logs as part of the inventory\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/2a523a8aa98dd4d152b7961ee17e95069096e39d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-10-11 16:51:02.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2023-10-30 23:44:02.000000000","tz":0},"subject":"Make sure we eject media from DVD when CD is requested","message":"Make sure we eject media from DVD when CD is requested\n\nIt\u0027s possible to use virtual media based provisioning on\nservers that only support DVD MediaTypes and do not support CD\nMediaTypes. The problem in this scenario is that Ironic will keep\nthe media attached since it will only eject the ones matching the\nCD device, now we check if there is any DVD device with media inserted\nwhen looking for CD devices.\n\nCloses-Bug: 2039042\nChange-Id: I7a5e871133300fea8a77ad5bfd9a0b045c24c201\n(cherry picked from commit 766d2804a1743e238a37762c946dec0984721167)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/235e6ccb9835a2583984a1a47614b8bab1dbad0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/235e6ccb9835a2583984a1a47614b8bab1dbad0e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
