)]}'
{"id":"openstack%2Fironic~899337","triplet_id":"openstack%2Fironic~bugfix%2F22.0~I7a5e871133300fea8a77ad5bfd9a0b045c24c201","project":"openstack/ironic","branch":"bugfix/22.0","topic":"handle-dvd-eject-bugfix/22.0","attention_set":{},"removed_from_attention_set":{"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2024-11-25 14:33:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7a5e871133300fea8a77ad5bfd9a0b045c24c201","subject":"Make sure we eject media from DVD when CD is requested","status":"ABANDONED","created":"2023-10-30 23:44:37.000000000","updated":"2024-11-25 14:33:55.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":897989,"cherry_pick_of_patch_set":3,"meta_rev_id":"bd295f6fd98d905818222d63eb05245d1e8672f1","_number":899337,"virtual_id_number":899337,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-05 12:03:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-10-31 08:04:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-10-31 08:04:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 23:54:27.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-10-31 00:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-31 08:04:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-11-10 11:11:48.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"}],"messages":[{"id":"f125758e0704d30214c66966d463b626a204005e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-10-30 23:44:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"caab4c30180c7a4b86eb557e65083f78c585b78b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-10-30 23:54:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-virtual-media-direct finger://localhost.localdomain/04b749e2d7cc4d5ea6b46f63581b9063 : RETRY_LIMIT in 10s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-ipxe-direct finger://localhost.localdomain/68f02ba801df4381865791cbf0b5c95c : RETRY_LIMIT in 5s\n- HPE-Proliant-ironic-container-standalone-wholedisk-ilo-uefi-https-direct finger://localhost.localdomain/0ce18ba689e04c18b76b6e04bc75f9b0 : RETRY_LIMIT in 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ba33751656e872ac561d9ccd7aa034f0c753e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 00:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b319aa37eb64b98b46450bfc1152bda\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0392f52b0ae04aedbaaa792672f5a8f5 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/682d6803a102445eb0965f56ddbdacd6 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee2a1c6b29f34350aa5e61d0c4ff6741 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c47408fa6e4fe2b7d6c38a95e41316 : SUCCESS in 15m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9876162f53084cb29ffbd860ecfb7eee : SUCCESS in 9m 01s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/d2d162f1cd614190ab2b9e7b890c2ca3 : SUCCESS in 10m 17s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e0791a47d28f44528cac7f94939f44cc : SUCCESS in 8m 16s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/9f0a06857fe24c9aae74ddee9adabac9 : SUCCESS in 40m 43s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/5ae99c2b33dd4db7aa92167dca2cf0a7 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"603de3fda0493795054a80a4b0d8e7a8717e1c6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 00:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49459f9e3fe4e51be59ede9df09c199\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35a3712813fe4557a2ff44091ded541d : TIMED_OUT in 1h 00m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42e0262191254366beb24539fc317d54 : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"469ee9b75c905190de950cb7567c31e4a5abb551","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-31 08:04:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed171f7eac1d80dfa38a6ca579347d17a7aab64c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 08:05:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cfa9040e42644fa1950b2a7c1c933b6bdf909c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:50:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b762fd3146d944e1be1f364704daa8ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4231fe17eb9946aaa012f950f535881b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e006423578f04842bbb61339ad141f4b : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee804d7111c46bfada86e5304048c84 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67da04a19d404c6b986e817c7c84e0df : SUCCESS in 9m 03s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/e8b1f4dea2a84153a2495ab40eedc4c1 : TIMED_OUT in 30m 49s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/fef2d76316ce4748805239715fc3658c : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"82fd8edcabd26aacbc84cdba7616ca0063168768","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2024-02-05 11:08:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"119d8394522cd0faa4c73fcb69f610e3b6b13632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:17:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caaa9719b1e2477fac2ecd726acd6f30\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9263b59e53a04e2897e72200ec91b328 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ce00ad650444af5ac7d3400c5aa95a6 : FAILURE in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5df4002f5080a4bdeb9fbd058e3faf4599b0649b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 12:03:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ecaf442f96a4ee4a37b8de28f103db4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f161dcaddc43129305e435393af9bc : FAILURE in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69620f96d261412c90686ed5cf50fd13 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72fbf0f5dfa043a99b521b303dc2f394 : FAILURE in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991751db4f5042f3ab477652110b4896 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/843633c5d0f34f32b1e350ae449d8f9a : SUCCESS in 6m 46s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/ae91496f88f246c0b117e76ba4330c3c : FAILURE in 11m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/162c696640dc460cb14df7d36c851e22 : FAILURE in 7m 22s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/a378b727cc3f48fa8e16438e1b06543c : SUCCESS in 45m 07s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/28a09b9472f349d0b1040a9e6edb0aff : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2193be45c3101eee5936e924bf0095d57ba7ecb6","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-11-10 11:11:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c878d9999be865827e3d7f48436473214252dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-10 13:07:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b9eb462fd9249fc8cae5771b32744ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5fcf69999074f55b23588581a28ae5f : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b7be73e01e84b9e81907654e9a73610 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9637d13646a4dec9dc6f8b107283352 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59deafd0ec7e4c71baf7efbcb8e7062e : SUCCESS in 15m 44s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/44d75962149b4cc5bb4eebedda4f93c1 : FAILURE in 8m 56s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/42b7688dc0ac4868aa9051b8e5d2cfb2 : FAILURE in 7m 01s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/7857f4b9fb6642b7b365842a434fac4e : FAILURE in 1h 55m 10s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/ad25c3fab5fa466bacec0cbf19a2ee96 : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e78f9627549491009f164185a656b4a79b6b36a7","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-11-11 00:02:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39db8a3bcaacb93624cb37e70c4d5d510555b77c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 01:49:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12a5a2aabf7e49189502908914dbbe63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd5310b562324786b6d542383f8e8e9e : FAILURE in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb93feb9d1784d65ac2145777caca17e : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ea2720b6ce499fbe98519ef7c819b6 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ead5afe9b004d3d8296c449a40650ca : SUCCESS in 12m 07s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/5d6120a7dee94e75b78597d8c70f04c3 : FAILURE in 11m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/0c88790b1ab04251abbf4ed648372bea : FAILURE in 9m 14s (non-voting)\n- metal3-integration https://zuul.opendev.org/t/openstack/build/8bc69431bec745b08c9247b153b9e2fb : FAILURE in 1h 46m 29s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0d21a3a9166c4015b596b1b68ab0e03a : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43406bfb895e28ddabe0acef2bfedaa7b87da4f3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-11-25 14:33:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd295f6fd98d905818222d63eb05245d1e8672f1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-11-25 14:33:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8988e3616e9e693d7c5a2cb79f076eec9968e84","revisions":{"f8988e3616e9e693d7c5a2cb79f076eec9968e84":{"kind":"REWORK","_number":1,"created":"2023-10-30 23:44:37.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/37/899337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/37/899337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/899337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/899337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/37/899337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/37/899337/1"}}},"commit":{"parents":[{"commit":"5b3dffdaf2be3443e11137b924609c84434fdc6e","subject":"Fix the HTTP code for reaching max_concurrent_deploy: 503 instead of 500","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/5b3dffdaf2be3443e11137b924609c84434fdc6e"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-10-11 16:51:02.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2023-10-30 23:44:37.000000000","tz":0},"subject":"Make sure we eject media from DVD when CD is requested","message":"Make sure we eject media from DVD when CD is requested\n\nIt\u0027s possible to use virtual media based provisioning on\nservers that only support DVD MediaTypes and do not support CD\nMediaTypes. The problem in this scenario is that Ironic will keep\nthe media attached since it will only eject the ones matching the\nCD device, now we check if there is any DVD device with media inserted\nwhen looking for CD devices.\n\nCloses-Bug: 2039042\nChange-Id: I7a5e871133300fea8a77ad5bfd9a0b045c24c201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f8988e3616e9e693d7c5a2cb79f076eec9968e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f8988e3616e9e693d7c5a2cb79f076eec9968e84"}]},"branch":"refs/heads/bugfix/22.0"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
