)]}'
{"id":"openstack%2Fironic~906113","triplet_id":"openstack%2Fironic~master~I1bf2bb6aba1a60de8448ea3fea011d8fdb815a45","project":"openstack/ironic","branch":"master","topic":"linuxefi","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-02-21 11:13:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1bf2bb6aba1a60de8448ea3fea011d8fdb815a45","subject":"DNM try s/linuxefi/linux/","status":"ABANDONED","created":"2024-01-19 14:01:03.000000000","updated":"2024-02-21 11:13:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ee6c3e9f2f23e95e15f4bf03c88d286f39bc23c","_number":906113,"virtual_id_number":906113,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-19 15:09:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-19 14:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-19 15:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"746f0852da2ef5a35317cbad456db333bda507a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-01-19 14:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77a8337f7ca69c55b2f9fb7569b41d9591858d30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 14:11:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d79fcfa70af0480c90ef28cc5b789975\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32c74ed4b7754fcc8e867f934f33b366 : FAILURE in 9m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe15b24da42b4097922a8695ce7c7441 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/460ccf7893c54485a53b7096af9cb5de : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b31cf090815d7c7e369d4662ada1b4ddad045ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:09:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be97b7f0ecbe42e387296eb34e7af2c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb89684b4c3b496d931bb8d99ec2ae92 : FAILURE in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6c32fa51905441bbfbc4571d9fee20a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b24cf08976a4e379e9b9fd74adf51e3 : FAILURE in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1a9698a6b634519ba0e562ca3c47981 : FAILURE in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cd0cb01b4444fda60957615e58d6fd : SUCCESS in 13m 46s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/cdf12f6f565f482aa6b0590b98853a10 : SUCCESS in 50m 15s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/93863893c25e4eabb32a06f1a6a126a6 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1ee6c3e9f2f23e95e15f4bf03c88d286f39bc23c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-02-21 11:13:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee18c35f4120597980db3684fc1973356a7d380d","revisions":{"ee18c35f4120597980db3684fc1973356a7d380d":{"kind":"REWORK","_number":1,"created":"2024-01-19 14:01:03.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/13/906113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/13/906113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/906113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/906113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/13/906113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/13/906113/1"}}},"commit":{"parents":[{"commit":"a374a0c123be83819a605391352ee7f66da35a7b","subject":"Merge \"Validate [deploy] image_server_auth_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/a374a0c123be83819a605391352ee7f66da35a7b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-01-19 14:00:49.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-01-19 14:00:49.000000000","tz":60},"subject":"DNM try s/linuxefi/linux/","message":"DNM try s/linuxefi/linux/\n\nChange-Id: I1bf2bb6aba1a60de8448ea3fea011d8fdb815a45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee18c35f4120597980db3684fc1973356a7d380d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/ee18c35f4120597980db3684fc1973356a7d380d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
