)]}'
{"id":"openstack%2Fironic~909918","triplet_id":"openstack%2Fironic~master~I2c6c82835e75cf29c3b50ef769ffa59033cdfa73","project":"openstack/ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-03-27 21:05:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2c6c82835e75cf29c3b50ef769ffa59033cdfa73","subject":"WIP/DNM: Don\u0027t setup fake first!!!","status":"ABANDONED","created":"2024-02-22 17:59:23.000000000","updated":"2024-03-27 21:05:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e14e3cdcb1d85a72e59c27f99ba94e2e1d7af624","_number":909918,"virtual_id_number":909918,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-23 02:52:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 18:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 20:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6eeb6ae43fe74862be6d0f7a3d8358071bfa0b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-02-22 17:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24cc2ec3d41e948be81a0a3a36b83cc76209c5f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:18:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49c178999c04b5eaf7e13700ebfa3ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d178eac64af74fe6a23464b7699f6e89 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45c401b8e93f419cac38aa2aa17bd1b9 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/399674820e3444c9ace1da0b45335358 : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cdfd71f17c822d496e733d06283be9364a929f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:08:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80592af576249778ecf59dbebd1d6d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5b40b407bff4f6eb02f50fd4f97ccd6 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6086e9f4cae94ec8ae50a680f3aef67d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c8e865a523943a9a21cfab899e22a61 : SUCCESS in 9m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14922c5c64bd4cf0952d20494061196c : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a64cc253554c7199abd63c9f3971de : SUCCESS in 14m 06s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/eeec431c5f094d2fb714672f44640781 : SUCCESS in 5m 05s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0cb73bb0685642439b4af794e5f07fc9 : SUCCESS in 5m 05s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4a0be91f17334bc29c0951c9d8a50059 : SUCCESS in 10m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c8cdd4666ebe4482a3a7b79dcb231843 : SUCCESS in 8m 50s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e8eed4119985426ebb03c7ac5bda124d : SUCCESS in 8m 03s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/778f17476ec945f09ad290abc65c8acd : SUCCESS in 27m 41s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/a50dabd6cf6f4552a7871710a375338d : SUCCESS in 32m 44s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/83065c6ae24143ec934c8976263d1517 : SUCCESS in 57m 32s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/496ddb48a8f349f3a759e32320982c65 : SUCCESS in 58m 29s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/b55d08046e904040816b2064fbdf1218 : SUCCESS in 1h 07m 39s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/dfe35ba5fa2a480b80fd7a898edf1581 : SUCCESS in 51m 23s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/144469fc3057417bada6e620e31a9b7f : SUCCESS in 47m 39s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/90d728306a3c41d4a73a7e3726509d06 : SUCCESS in 50m 13s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/e079d5e0ef1f41c99b5e0068a2e2a254 : SUCCESS in 31m 09s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/f4173cad81f14e4bbea51e47ab2cb2b8 : SUCCESS in 59m 17s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4b552ce1c3984f94a617af724b284cb8 : SUCCESS in 1h 57m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f2ecdcbf8c1047e6b5d2f5fad63878f4 : SUCCESS in 53m 57s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/be44d14f7dc14b909b87dbc47143f91b : SUCCESS in 43m 16s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/aad95950a77a48cb90c3cfd82153dc30 : SUCCESS in 58m 28s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/8b02ce4aafeb4bfc9973a0f7c5f606ea : SUCCESS in 1h 04m 04s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/42e9955713284c0db03a88938fc3217f : SUCCESS in 38m 42s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1faf59b171a048199107c4458c1e8cf4 : SUCCESS in 49m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/84e38f2b60844d88aee2e10f99370360 : SUCCESS in 40m 36s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/44bef35ae3454e62afcd9a326e5d6a4f : SUCCESS in 1h 01m 01s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/dcf9595183c44a6e861eda973b61bb64 : SUCCESS in 52m 21s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4dcf801ac46243f4a2f581cc38557277 : SUCCESS in 58m 05s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba40b24ad8a243249a9abb238911d03b : SUCCESS in 45m 26s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/da4a485182cc47f28c2118e88b7842e0 : SUCCESS in 40m 55s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/f6d23e2e396e4589a5b57ca98e63b0e8 : SUCCESS in 1h 05m 07s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/bd4bbca6498c42248c6ed6d09908eb30 : SUCCESS in 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ceeb5cd11e5cf4da5a131dab1bf8d9f514da82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-02-22 22:11:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b97ef26fc2393a8b3415313b17678ea21304d11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 22:21:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1bbcb63dcb4987aaea4dab5524dffc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2474335b237410299c8166557936f1e : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8856b729dd4d414eb935dbe37ad1d999 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9fcaa50ca814d79a6c6d8be422ec6a3 : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9433a4ac4c3e6fbec936f7eb50839ffc58dcfcee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 01:19:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b101f801f2c64b9399055b352868009c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65701a247e8942d2ac4bd6ab1e14a576 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ce4ac504a64bf29e06bc7cd1fb736a : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8741b2dc79a485ba11c634ce43de2aa : SUCCESS in 7m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c40b24c9f7cc44c0a6397fd079fe5b16 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc377c56ba04d2aa0ad0f6578d0e49c : SUCCESS in 14m 12s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/99e8748eb99440c5b4f592161baae2ad : SUCCESS in 4m 36s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/b08d32d3b8e745e9ba804ece56687b5c : SUCCESS in 5m 48s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/283442a2f4984d50bd8162ef5dc74703 : SUCCESS in 9m 37s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/f8930ff98f364905829b38b328a4a028 : SUCCESS in 7m 32s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c79d55f9674b4e19804a0157508b834d : SUCCESS in 8m 42s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/e9a4e6a3b296413c9a2589480042ad4c : TIMED_OUT in 3h 02m 03s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/e0d643ff6c0f4592a8794c224cc93957 : TIMED_OUT in 3h 02m 13s (non-voting)\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6f05915305ad40fb9d4a0789a88408bd : SUCCESS in 1h 10m 34s\n- ironic-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9104f4f1da08435093e9b3c208405147 : SUCCESS in 1h 11m 51s (non-voting)\n- ironic-standalone https://zuul.opendev.org/t/openstack/build/c9baa3937db64143b70583aaf24d180c : SUCCESS in 41m 33s\n- ironic-standalone-redfish https://zuul.opendev.org/t/openstack/build/e6bd6fc82600468abe575bd626275768 : SUCCESS in 1h 12m 29s\n- ironic-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/22d64cc28a154777917b70f3f844661c : SUCCESS in 29m 00s\n- ironic-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ea88ae16feed4f429acca098ba3be736 : SUCCESS in 39m 03s\n- ironic-tempest-ramdisk-bios-snmp-pxe https://zuul.opendev.org/t/openstack/build/214c9335389648ca93c8699ebd979ea4 : SUCCESS in 37m 05s\n- ironic-tempest-partition-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/00d37ca35caa49d99e3e28a026492add : SUCCESS in 59m 11s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f3c4b2fba9d84d66b1a97abf8a7f454d : SUCCESS in 1h 30m 15s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/75e85e5b0c3748ed94f69aad1826799f : SUCCESS in 1h 09m 22s\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/e2301ad3c8574b01a2a008f2233928fa : SUCCESS in 46m 34s\n- ironic-tempest-ipa-partition-uefi-pxe-grub2 https://zuul.opendev.org/t/openstack/build/404d7511c4fb43b6b13acd043f2be2d8 : SUCCESS in 58m 18s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/57e0a613a6654d1eb497b6c98405f282 : SUCCESS in 1h 06m 50s\n- metal3-integration https://zuul.opendev.org/t/openstack/build/d9b5680643474a2eaed13179b0c0ca00 : SUCCESS in 41m 18s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/877e8bcfb5af456895b9c158c4fb6172 : SUCCESS in 51m 57s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-ipmi-direct-dib https://zuul.opendev.org/t/openstack/build/66e497dc5ed146c3998b95f38ac3236e : SUCCESS in 42m 46s (non-voting)\n- ironic-tempest-ipxe-ipv6 https://zuul.opendev.org/t/openstack/build/c7bf1188fe9a43aca69f0d2a06e10f5c : SUCCESS in 42m 14s (non-voting)\n- ironic-standalone-anaconda https://zuul.opendev.org/t/openstack/build/d1689d0d05284e3cb547810511a36687 : SUCCESS in 33m 01s (non-voting)\n- ironic-inspector-tempest-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/f69ee2ad63cc4686bbae07a1ac7377b2 : SUCCESS in 39m 40s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9ded40c453d4de29b721f1e36a5e466 : SUCCESS in 31m 17s (non-voting)\n- bifrost-integration-redfish-vmedia-uefi-centos-9 https://zuul.opendev.org/t/openstack/build/eab9d21c8b0b4a9981332f173c949000 : SUCCESS in 34m 59s (non-voting)\n- ironic-tempest-pxe_ipmitool-postgres https://zuul.opendev.org/t/openstack/build/f590732ec3894534b8cc86c5ae79492b : SUCCESS in 38m 51s (non-voting)\n- bifrost-benchmark-ironic https://zuul.opendev.org/t/openstack/build/8b85f5be766c49c1b678226b58f7aebf : SUCCESS in 51m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"652d14cc4104149fa5c0b1703f6cb8bd9bbc6b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-02-23 01:30:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5c73244599f089fdc60078e5e2b84a198181d0b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 01:48:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b532b06383774309a6bbd73b8a102ded\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d69f48fa0b04265958882ae4d624707 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c78edd1c5ef14c998bda5db34bd37702 : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7fde5b5656b4d5dab252ab8e37d80ea : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f259e8247b28e568f28bdf54322c5be533dbba2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 02:52:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c00c85e5003d41ff90a32f9a2843e4db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2f8e7345f0f4692ae2fe4556cd44ed9 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd96a954d07474bbf26e0d279035be9 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76a5db0e8a834edf86cced3e5597a0ed : SUCCESS in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e64eecfd2b14bdfbbf938175cff1f79 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5ddf1dc457471e823f77a18360af16 : SUCCESS in 11m 01s\n- ironic-tox-bandit https://zuul.opendev.org/t/openstack/build/0d900a4fe14b46b685cdcd21ee980d47 : SUCCESS in 4m 50s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/f22e218422ff48e1a7f39688e63e0b85 : SUCCESS in 5m 35s\n- ironic-tox-unit-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/980debf8f5df447abead1c503b92bb9b : SUCCESS in 7m 54s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/478c7680e5a049658f2d3c0439a2dc8c : SUCCESS in 7m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/10713f75f7a34e65b1fe86f5ebeebb87 : SUCCESS in 6m 05s (non-voting)\n- ironic-tempest-functional-python3 https://zuul.opendev.org/t/openstack/build/ccfd97dcd60d417a94add3a9efa3a27a : FAILURE in 1h 14m 53s\n- ironic-tempest-functional-rbac-scope-enforced https://zuul.opendev.org/t/openstack/build/4cb619cce9fb403eb9124041a041f556 : FAILURE in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e14e3cdcb1d85a72e59c27f99ba94e2e1d7af624","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-03-27 21:05:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ed92a67720de9d718e3145492baf6ecefa4c25c","revisions":{"f34481b49c503da7bb6882b860691bab36510cd7":{"kind":"REWORK","_number":1,"created":"2024-02-22 17:59:23.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/18/909918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/18/909918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/18/909918/1"}}},"commit":{"parents":[{"commit":"0cfe290d1508b12e8797354363f739fe27df4e53","subject":"Merge \"Ensure all errors are passed during cleaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0cfe290d1508b12e8797354363f739fe27df4e53"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-22 17:59:00.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-22 17:59:00.000000000","tz":-480},"subject":"WIP/DNM: See what explodes when cleaning is enabled for functional tests","message":"WIP/DNM: See what explodes when cleaning is enabled for functional tests\n\nChange-Id: I2c6c82835e75cf29c3b50ef769ffa59033cdfa73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f34481b49c503da7bb6882b860691bab36510cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/f34481b49c503da7bb6882b860691bab36510cd7"}]},"branch":"refs/heads/master"},"0d8feefc57ed533049fc0c9a9169af747aaa643e":{"kind":"REWORK","_number":2,"created":"2024-02-22 22:11:57.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/18/909918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/18/909918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/18/909918/2"}}},"commit":{"parents":[{"commit":"0cfe290d1508b12e8797354363f739fe27df4e53","subject":"Merge \"Ensure all errors are passed during cleaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0cfe290d1508b12e8797354363f739fe27df4e53"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-22 17:59:00.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-22 22:11:49.000000000","tz":-480},"subject":"WIP/DNM: Don\u0027t setup fake first!!!","message":"WIP/DNM: Don\u0027t setup fake first!!!\n\nChange-Id: I2c6c82835e75cf29c3b50ef769ffa59033cdfa73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d8feefc57ed533049fc0c9a9169af747aaa643e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0d8feefc57ed533049fc0c9a9169af747aaa643e"}]},"branch":"refs/heads/master"},"6ed92a67720de9d718e3145492baf6ecefa4c25c":{"kind":"REWORK","_number":3,"created":"2024-02-23 01:30:36.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/18/909918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/18/909918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/18/909918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/18/909918/3"}}},"commit":{"parents":[{"commit":"0cfe290d1508b12e8797354363f739fe27df4e53","subject":"Merge \"Ensure all errors are passed during cleaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/0cfe290d1508b12e8797354363f739fe27df4e53"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-22 17:59:00.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-02-23 01:30:30.000000000","tz":-480},"subject":"WIP/DNM: Don\u0027t setup fake first!!!","message":"WIP/DNM: Don\u0027t setup fake first!!!\n\nChange-Id: I2c6c82835e75cf29c3b50ef769ffa59033cdfa73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6ed92a67720de9d718e3145492baf6ecefa4c25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/6ed92a67720de9d718e3145492baf6ecefa4c25c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
