)]}'
{"id":"openstack%2Fironic~928049","triplet_id":"openstack%2Fironic~bugfix%2F25.0~If04a5b97722cc1a8e125c3348e09339c3a7ce0eb","project":"openstack/ironic","branch":"bugfix/25.0","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-09-04 18:33:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If04a5b97722cc1a8e125c3348e09339c3a7ce0eb","subject":"CI: Disable metal3-integration test job","status":"MERGED","created":"2024-09-04 17:49:40.000000000","updated":"2024-09-04 18:37:05.000000000","submitted":"2024-09-04 18:33:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928049","cherry_pick_of_change":928047,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a75f1bc6cc81f545c1fc0ae4e599201f537d3d9","_number":928049,"virtual_id_number":928049,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 18:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-09-04 17:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-04 17:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-04 17:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 17:55:18.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-09-04 17:55:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-09-04 18:15:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15b082e79b7708ab071f3e64ec6f461e58eddd3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-09-04 17:49:40.000000000","message":"Patch Set 1: Cherry Picked from branch bugfix/26.0.","accounts_in_message":[],"_revision_number":1},{"id":"74aed4d936c317345f863fb07b94daf1b8fc2f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-09-04 17:55:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"866c91f31cb60183ba8802b0356e2cc4142dbc5c","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-09-04 17:55:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4bfb2872603ea84da6d290a33714013ab7a9bf89","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-09-04 17:55:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3710e77a94909a0ab2bd4ba1e3989eac14d45a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:15:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e374bef3f84f1ba9474dfffe13caee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/726e95c5d1624224b13f782fde7c7300 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e05e3593dc974a61bc95cc2f55ce6b95 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d32e922b9a443e8b61184945c6ebdf7 : SUCCESS in 14m 19s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/589b3799cdb84e8589c7668ea2992456 : SUCCESS in 3m 55s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/0f75ca2a616d439e967424e19885c24c : SUCCESS in 4m 51s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/c2da591fa793439baed902bde3c4d0af : SUCCESS in 8m 07s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/424639885b0f4f73af7faaa73b02316b : SUCCESS in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa0d4fbea76a6804f89a53e3976471d760d62599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:16:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b326b08e45591d475465cb13d67a03364a57276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:33:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2652151a5a4b4947988519ba252a44fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634ba3edd4924f4faa1e4ada3983d59d : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc43d6b7440f46cf8d155831f3bdbc37 : SUCCESS in 10m 55s\n- ironic-tox-codespell https://zuul.opendev.org/t/openstack/build/aaf66e9285a3447a87cf7b50e55f2259 : SUCCESS in 3m 47s\n- ironic-tox-unit-mysql-migrations https://zuul.opendev.org/t/openstack/build/8858cbf3b1954ecfba4554d0967aa494 : SUCCESS in 5m 02s\n- ironic-tox-unit-with-driver-libs https://zuul.opendev.org/t/openstack/build/97e47a425c53405ebc4570d4690bb8c4 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d00bc4881c56620b22f1761e54b2bf8f7f99bd01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:33:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4a75f1bc6cc81f545c1fc0ae4e599201f537d3d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:37:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a6d3db2d35486b9bf847ca99581e5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101124e9307c4d5fb97f30e510d142ba : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4cb0af7fd6e9afa8fe8a3b2a2e47427929068843","revisions":{"07caca02d19cc7cb88e5c6d402210d031e0b03c4":{"kind":"REWORK","_number":1,"created":"2024-09-04 17:49:40.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/49/928049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/49/928049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/49/928049/1"}}},"commit":{"parents":[{"commit":"b7778961d40a0590ce270887bf81a9314c84c0dd","subject":"Update .gitreview for bugfix/25.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b7778961d40a0590ce270887bf81a9314c84c0dd"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-09-04 17:46:50.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-09-04 17:49:40.000000000","tz":0},"subject":"CI: Disable metal3-integration test job","message":"CI: Disable metal3-integration test job\n\nThe metal3-integration CI job is not smart enough to know which\nbranches to pull for it to correctly test the branch, and so it\nshould be disabled on this branch.\n\nChange-Id: If04a5b97722cc1a8e125c3348e09339c3a7ce0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/07caca02d19cc7cb88e5c6d402210d031e0b03c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/07caca02d19cc7cb88e5c6d402210d031e0b03c4"}]},"branch":"refs/heads/bugfix/25.0"},"4cb0af7fd6e9afa8fe8a3b2a2e47427929068843":{"kind":"REWORK","_number":2,"created":"2024-09-04 17:55:02.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/49/928049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic","ref":"refs/changes/49/928049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic refs/changes/49/928049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic refs/changes/49/928049/2"}}},"commit":{"parents":[{"commit":"b7778961d40a0590ce270887bf81a9314c84c0dd","subject":"Update .gitreview for bugfix/25.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/b7778961d40a0590ce270887bf81a9314c84c0dd"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-09-04 17:46:50.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-09-04 17:54:55.000000000","tz":-420},"subject":"CI: Disable metal3-integration test job","message":"CI: Disable metal3-integration test job\n\nThe metal3-integration CI job is not smart enough to know which\nbranches to pull for it to correctly test the branch, and so it\nshould be disabled on this branch.\n\nChange-Id: If04a5b97722cc1a8e125c3348e09339c3a7ce0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4cb0af7fd6e9afa8fe8a3b2a2e47427929068843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic/commit/4cb0af7fd6e9afa8fe8a3b2a2e47427929068843"}]},"branch":"refs/heads/bugfix/25.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
